)]}'
{"id":"openstack%2Fneutron-lib~804203","triplet_id":"openstack%2Fneutron-lib~master~I8b18785f37847804dd84e810544e95538d1a56e1","project":"openstack/neutron-lib","branch":"master","topic":"bp/qos-minimum-guaranteed-packet-rate","hashtags":[],"change_id":"I8b18785f37847804dd84e810544e95538d1a56e1","subject":"Add port-binding-allocation-groups shim API extension","status":"ABANDONED","created":"2021-08-11 06:23:51.000000000","updated":"2021-08-16 08:45:55.000000000","total_comment_count":6,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"c6afb882d6b1141ffe8c262a6711a8263d73f083","_number":804203,"virtual_id_number":804203,"owner":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-12 12:12:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-11 06:25:36.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-08-11 06:25:36.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2021-08-11 06:25:36.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-08-11 06:25:36.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-08-11 06:25:36.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-11 07:51:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"000dba34831aa725a588aa0ad5cb38d5417569fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-11 06:23:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c196e88dabd0c1be38215c88cc2ee31fe0ce177","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-11 07:43:44.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6adf2cbae4a61f27c76700837dc6a161ec8d2b5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 07:51:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7fba8d06a4d4217a4d7e9376a80ad0d : SUCCESS in 9m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/692175f2e2e347f1b51959808afc7dfa : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f1665bedf794aabbfe95ae65e457a4b : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dff07051c224644888e42aa23702c70 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1f40cd7b7c143f0ab4f1dea8e909d36 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4550e2d1b214945aa1a8a3265dc863b : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b54aaa592e794693a82970d7f710b045 : SUCCESS in 7m 44s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5bc90cda8ea64f48a5470f12b9c01198 : SUCCESS in 1h 23m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99420448b4304f7b82bd300b12ab63cf : SUCCESS in 7m 05s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3763f182365a4270a1e1d4fcc676a734 : SUCCESS in 52m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a6c8316625946aab1784145fa5da8ec : FAILURE in 46m 41s","accounts_in_message":[],"_revision_number":1},{"id":"aa11cd2d00bc76d3e566d8ecc1ad34d397ed9292","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-11 07:53:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"13990aeccbd3b49f29a2adbefb6aaba3cc153c80","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-11 07:59:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"818c19ba1ff46a16f379631f6ab023a625aa12b0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-11 08:00:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"646e3917417f3062e5e754bd21f6109958ab8012","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-11 08:08:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de930062475eb948df08fc9b64f9dfe7e42b8f04","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-11 08:15:29.000000000","message":"Patch Set 1: -Workflow Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"94533d59b4b41ed43287a26193a6d8d893f3b4d9","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-11 08:47:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d69d1edd9cca73745ccf6aac9e594a184449d062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 09:16:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/322f16b41fcd4b07963047e4523ac4fc : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d1a802db6194d80905dc125f85ac4bb : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff03c5e22b0346ba810da06f9170df34 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6ab52b51f3844a19038563f9d628427 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7beb51185b8f4f1b886ecee6101ebd62 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a730d0fa38454aceb14b2b80112cae46 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3151b0d7aa8240f18479cc97a6130dfc : SUCCESS in 5m 40s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebedbab1c38447069ca931b6ab046dc5 : SUCCESS in 1h 21m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83b2fda1478a4f62a609c5b1d6bab0ef : SUCCESS in 4m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/58594f71010a4081937cf4c46b60a91a : SUCCESS in 43m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5beabe2cf5bc4ba0a0977eb5677eb885 : FAILURE in 46m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b6b21bff3695b9b3386061f165c2414969332e8d","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-11 10:24:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"78c1a72cf7cfb3c4e506c835e28df57997b74b88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 11:45:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf50c9b50cda45268dacd640db0f42ff : SUCCESS in 7m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d893e189e4f24145a9c307206f74b5fc : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51680d0c09534366adccbd50b34390ee : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c16e7b16e2441eb418c86c9d1f3d39 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c60a28cbc5c43e3aad7ffa4177acb7e : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a936642ce5142a587680825b96db2bd : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29502c4f4c1f427bab75044acfe2acbf : SUCCESS in 6m 22s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0e2a3beb0fc4022854d2454285cbbd4 : SUCCESS in 1h 15m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39b6e190c6ba409da040c358834c8279 : SUCCESS in 3m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/545e933e6c8a4482ba42ff82fc007bfa : SUCCESS in 45m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/873e9395ad184631a6b28753d4279e7f : FAILURE in 45m 39s","accounts_in_message":[],"_revision_number":1},{"id":"cfac5e5e594b53604f54fc0fdc47ae62469f2887","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-11 14:00:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d5a1a8cc01dba92a08459f5ea002d327731a59da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 15:22:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8ce4de806e46cebcecb7ade8caa057 : SUCCESS in 7m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33105be7f9824885b137eedc99985ba5 : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41d1ee9e0d8d46c58beeb714265fde63 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70caf739976b48e8bae85850f577d273 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33caf1f8a2414320abef60784294e0c0 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23020251df9748a1966d6ad218389f6b : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c136b97cc714bdbbe47a2b4cdc7feea : SUCCESS in 4m 26s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1865d87e6a9d4a75b92787b5ae1be5e9 : SUCCESS in 1h 20m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcee600432bb4a0ba70043d132167161 : SUCCESS in 4m 51s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a38599ddff584be2bdfa82df83306b2d : SUCCESS in 45m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/84763f0f008c4c1696ed50408860b1f5 : FAILURE in 43m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5a9786a4708dbe9a5971d0d3ad784140545a9ec1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-12 10:49:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f86e487788a984bb91d72e15fdb1d1b9d9739f3","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-12 10:52:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84c9c1355d9c3dcfd5fb61cfdf561f81956f9182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 12:12:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/952cc046ad454178b204f6b592ed58c9 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4ac4699406c40289ecb01a880fd385c : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/baaf644e20734b7d9ca74904672311a7 : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10986ba8048b4d78ab629e698f90692f : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3afc21b3ed2e4d878ed795ad1a03e3f0 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06ef1775571d470bb26e411bc75706e9 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/733a196d87ef4427bce2b1b4c62e6138 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4de0e061ef14ec481cbb3e86c1edae3 : SUCCESS in 4m 36s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c406aa4e3de47e4b2f187959e93d0ed : SUCCESS in 1h 21m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4840a7867bbd4060b13ae52d6b6d337a : SUCCESS in 4m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4b7f1cf592d141498e38fd091486f8ab : SUCCESS in 42m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9cb89dc5ba8c421b851280e3e6211908 : SUCCESS in 46m 48s","accounts_in_message":[],"_revision_number":2},{"id":"e90f69c25941a2e404c4a62a2ea2922e2b3ee542","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-16 07:07:28.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nDo we need to have two separate extensions one for the resource_request with groups and one for the allocation with groups? I\u0027m not against it just wondering if we ever use the allocation extension without the resource_request extension.","accounts_in_message":[],"_revision_number":2},{"id":"601b52ea15e8f113b06dc82d2ab7474dd965156b","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-16 07:16:47.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Do we need to have two separate extensions one for the resource_request with groups and one for the allocation with groups? I\u0027m not against it just wondering if we ever use the allocation extension without the resource_request extension.\n\nI don\u0027t think those extensions will ever be used separately, as they depend on each other anyways.\nI like the idea of keeping it simple and adding only 1 shim extension.","accounts_in_message":[],"_revision_number":2},{"id":"c6afb882d6b1141ffe8c262a6711a8263d73f083","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-16 08:45:55.000000000","message":"Abandoned\n\nAfter a discussion with @Lajos, @Bence and @Balazs we agreed that additional extension is not needed. If port-resource-request-groups extension is supported, it already implies that Nova has to use a new format of \u0027allocation\u0027 key in the \u0027binding:profile\u0027.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0c87d743e153eb7979a8b18664d8fb04bdd383cd","revisions":{"4890b96d950660c5d6fd1884be0fe2b84b65f133":{"kind":"REWORK","_number":1,"created":"2021-08-11 06:23:51.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/03/804203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/03/804203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/804203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/804203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/804203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/03/804203/1"}}},"commit":{"parents":[{"commit":"8c473f4d325e44b50629c21113a8bf1b5060c8f1","subject":"Merge \"Add API extension definition for QoS minimum pps rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8c473f4d325e44b50629c21113a8bf1b5060c8f1"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-10 14:14:06.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-11 06:21:29.000000000","tz":120},"subject":"Add port-binding-allocation-groups shim API extension","message":"Add port-binding-allocation-groups shim API extension\n\nAdd a new shim API extension definition called\nport-binding-allocation-groups. This extension indicates that\nNeutron supports port binding for ports with two groups of resource\nrequests.\n\nBased on this extension, Nova will know that it can send the new format\nof \u0027allocation\u0027 key in the \u0027binding:profile\u0027.\n\nRelated-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: I8b18785f37847804dd84e810544e95538d1a56e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4890b96d950660c5d6fd1884be0fe2b84b65f133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4890b96d950660c5d6fd1884be0fe2b84b65f133"}]},"branch":"refs/heads/master"},"0c87d743e153eb7979a8b18664d8fb04bdd383cd":{"kind":"REWORK","_number":2,"created":"2021-08-12 10:49:40.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/03/804203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/03/804203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/804203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/804203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/804203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/03/804203/2"}}},"commit":{"parents":[{"commit":"8b2fbe6b64b038e0d0adc4423835b4605e0e4cea","subject":"Merge \"[Docs] Add api-ref for the address groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8b2fbe6b64b038e0d0adc4423835b4605e0e4cea"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-10 14:14:06.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-12 10:49:07.000000000","tz":120},"subject":"Add port-binding-allocation-groups shim API extension","message":"Add port-binding-allocation-groups shim API extension\n\nAdd a new shim API extension definition called\nport-binding-allocation-groups. This extension indicates that\nNeutron supports port binding for ports with two groups of resource\nrequests.\n\nBased on this extension, Nova will know that it can send the new format\nof \u0027allocation\u0027 key in the \u0027binding:profile\u0027.\n\nRelated-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: I8b18785f37847804dd84e810544e95538d1a56e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0c87d743e153eb7979a8b18664d8fb04bdd383cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0c87d743e153eb7979a8b18664d8fb04bdd383cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
