)]}'
{"id":"openstack%2Fneutron-lib~804378","triplet_id":"openstack%2Fneutron-lib~master~I295d8004efd563a5044631ec504d4eb9ae1be55d","project":"openstack/neutron-lib","branch":"master","topic":"packet_rate_limit","hashtags":[],"change_id":"I295d8004efd563a5044631ec504d4eb9ae1be55d","subject":"Move packet_rate_limit type to neutron-lib","status":"MERGED","created":"2021-08-12 10:40:14.000000000","updated":"2021-08-17 17:03:07.000000000","submitted":"2021-08-17 17:01:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804378-packet_rate_limit","meta_rev_id":"9aba8920f61c69bcc039c204a6452fc10e851e87","_number":804378,"virtual_id_number":804378,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-17 17:01:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-08-17 06:48:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-08-17 17:03:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-17 17:03:07.000000000","post_submit":true,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-08-17 06:48:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-12 11:05:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-08-12 12:08:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-13 07:54:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-08-13 08:03:09.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-08-17 06:48:45.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"3db4b1c4b481f4e47b48c94677bdea5b7798c0c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-12 10:40:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1f13672fc4d1b6bd0e3820b292f18bf1d0962bc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-12 11:05:02.000000000","message":"Patch Set 1:\n\nShouldn\u0027t you also move this?:\n- https://review.opendev.org/c/openstack/neutron/+/796363/15/neutron/extensions/qos_pps_rule.py","accounts_in_message":[],"_revision_number":1},{"id":"c627d398aa069221738ca89d5c08db1f43a1f088","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 12:08:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15a03388307f4fdcb67bbf803c2bd08a : SUCCESS in 7m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd385daf29cd4cbaa3f4af5b0640dc53 : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ae870f4a45e4868bd95f7aafe79e965 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5f2ef3c96574d58a093b62762bf7be4 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35cd3e6d333b4510bfa11b9e353226cf : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6da3c770886740ed9fcba5029978dd1a : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e529e67202b144cca974be218c9ef4e1 : SUCCESS in 5m 40s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ab7db3823774a1da207cad39f786694 : SUCCESS in 1h 23m 50s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6b9a2ae581534247aba0f24745b9b4a0 : SUCCESS in 39m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/33777887a1694047ad8b9e4659456a4d : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":1},{"id":"165a85f50dc89ccc9fdd1b2ab82dc2a46f1a4c87","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-13 01:19:50.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Shouldn\u0027t you also move this?:\n\u003e - https://review.opendev.org/c/openstack/neutron/+/796363/15/neutron/extensions/qos_pps_rule.py\n\nThere is no such \"get_resources\" function defined in neutron-lib. The extensions now are static declarative. This is a function with locgical belongs to the base class ExtensionDescriptor.","accounts_in_message":[],"_revision_number":1},{"id":"bcc75f88c71c7f8f45569fa30f0b55bbc325bd31","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-13 01:20:55.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Shouldn\u0027t you also move this?:\n\u003e \u003e - https://review.opendev.org/c/openstack/neutron/+/796363/15/neutron/extensions/qos_pps_rule.py\n\u003e \n\u003e There is no such \"get_resources\" function defined in neutron-lib. The extensions now are static declarative. This is a function with locgical belongs to the base class ExtensionDescriptor.\n\nOh... I see, there are some constants. Thanks, it should be moved as well.","accounts_in_message":[],"_revision_number":1},{"id":"ab019e55dfb23a12ca9179b01aa60d9cd4737620","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-13 01:32:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9d86f3cde76bdc0d18b41c60c0d79747f1121f74","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-13 01:34:12.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Shouldn\u0027t you also move this?:\n\u003e - https://review.opendev.org/c/openstack/neutron/+/796363/15/neutron/extensions/qos_pps_rule.py\n\nDone","accounts_in_message":[],"_revision_number":2},{"id":"43d760ad4c1c53904db3d6d17d929c80c10aa88a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 02:36:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b63a83979fab4705a2c4276de7fb752c : SUCCESS in 5m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b011944e2af147a2a4235f95695286f0 : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68317390fdfe4f07888d1de1bbf198ec : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a5fa3ea2a7448bea2496b0ddbb812b8 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/093573539ed44ae295e15f3f12478e9e : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bccd41d33c94a0381df1e7526464392 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dcbd551f1574bfb996bfa525a999a4e : SUCCESS in 4m 02s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a727f90926294153a43785f636c19750 : SUCCESS in 1h 02m 24s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9940ea63889649839019b5e6d2e3b4ab : SUCCESS in 32m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b29cb1bbb2994968957fe2c1dba1c4c2 : FAILURE in 44m 53s","accounts_in_message":[],"_revision_number":2},{"id":"560cdf35aac896b432d64d8ad37f85bc2db629ec","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-13 07:22:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d88952b5c0db59e2535e5ea263238c94fbf170a9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-13 07:54:09.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cd32b4b4124f719eb2364c12fcc342bd400338b8","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-08-13 08:03:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2f0fa98d4f7fcb13b23871afca07c8430f761706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 08:47:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa69bab3d6c94717a17e5849bba9030a : SUCCESS in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b7a71ee59ca470991b42d114821c239 : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d24130dde3f42019e6dadb877007b1f : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b86c7f0a073047b4a4eb3fb91fb5b560 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fb794cbde734a35b266a9b155639214 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0895d35ffb5043e68197c3d7767fc768 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e74fcd3c4a814c9fb969d287b1c01018 : SUCCESS in 4m 31s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa5036758cfa44f29636d348b1f4552d : SUCCESS in 1h 20m 13s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/68c919471d514b409bc47aa082a3e839 : SUCCESS in 43m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/df7266033b554697b10250e27e1b9ad5 : FAILURE in 53m 18s","accounts_in_message":[],"_revision_number":2},{"id":"ab96245093833ea86a538792162dfab7c9d535ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-17 05:57:45.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0707b621c4bb44274cb1fc348bbfa9399ed49688","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-17 06:48:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c6f6dd883847e942d7ff9d508cc8e97c567af7e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 07:44:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f470881b17fc4423a12f0451e5df924b : SUCCESS in 6m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83269ab7384344b6b43710d66ce68036 : SUCCESS in 15m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49003b4d2f26498ab69fbaa06243633b : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/316887cd6aff40908271fecec4b9fc78 : SUCCESS in 15m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fe2ccceb01c4db5afd371479397ebef : SUCCESS in 12m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9cfbe31380647529708ef148bab677e : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9efa2b6435e243dba3ebe29519e9dded : SUCCESS in 5m 04s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f639f3a918bd4276a471162e6f88e3bd : SUCCESS in 1h 44m 56s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3e4f8d545a754d3d975b3bf10b51417d : SUCCESS in 55m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d0a39b4a2898437abd2de306a31db86f : FAILURE in 1h 05m 48s","accounts_in_message":[],"_revision_number":3},{"id":"e27550b349a969948ccca8bfb9bbe74251b8c3bf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-17 09:12:06.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c8e42264463f3a3cec3df41c544ad335fb9191ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 10:34:14.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2fac6b7794c47b69ce69a9b9eb8aaa0 : SUCCESS in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc8233d5788946e9921fceb04ad177a9 : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd3afd0247834f54973b8175b0380f72 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a96da591110c4b4d9b370a62ac3866ee : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0044e7616584e57a6de98654e71236a : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a348683c346847d8afb49b9c2919347d : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52412450e023471d9023eccae4aa8f67 : SUCCESS in 5m 05s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6a08b5865b5427895694fc0869d4c00 : SUCCESS in 1h 20m 39s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/331f264f068746708a8cec8eb2d7d8b2 : SUCCESS in 39m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4caa3470a5343e68d9a7a5022240280 : FAILURE in 49m 37s","accounts_in_message":[],"_revision_number":3},{"id":"7282429355f57b24caafe7aee8132cbedf08455c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-17 14:05:25.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"074f82a6c42578f81c6e4532b34622cf296802c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 15:37:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f4e715b62b74669943e69e77a213a23 : SUCCESS in 5m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3c689a62c1744a7b7e153eeab40f930 : SUCCESS in 4m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd98670c0d1341228e487c3124e3d8d4 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40fcd3d8fad04ba6a8ba2355e6dd84c3 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8061e4d2d32e4eb0975147f3af1c7397 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f0e34f3597f42cd95293b574024f908 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23b720f4b2a2439b9c7595879845b3d4 : SUCCESS in 4m 28s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2323d4fd8e5c428898693b30f2aca8a2 : SUCCESS in 1h 25m 52s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d0358ebc604d4f89a078fe6225d93b4b : SUCCESS in 52m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/621f381f35d04ee499d1c424102cff1e : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":3},{"id":"80784c4cb21c26cee87598d7614a11058c8b6add","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 15:38:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fca6e24fc605b0eb6eb408d5e15dcddfa2a44d83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 17:01:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ae008f24094818b45f37b18df85cab : SUCCESS in 8m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35fd9864b67a494796e54b75c6ec3ff8 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c50b072f41e64631851f277aa4eeed3d : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c538f5e5a492401d9f398b31d756c865 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea4d2b35f62845458628b71df97f5eac : SUCCESS in 4m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a71e671b85ee443596f13dba906493ae : SUCCESS in 1h 17m 57s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/02514e8d0d82433f8aaae73e29e10a8f : SUCCESS in 44m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b1c6a34a45943898651153c5bb174bf : SUCCESS in 47m 59s","accounts_in_message":[],"_revision_number":3},{"id":"9967b904edb56cf831a729dead87595d60a8e929","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 17:01:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3cc0e40eb925d1ea6cc3e5138257b4cfcd10faae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 17:03:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a41b7eab4c2c4872a5a75d487ee016b4 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8ca7087ed978e412c4b31e543f3c31b824fbb779","revisions":{"cb209a112e13e0a8de9605173bac816e08e5a11e":{"kind":"REWORK","_number":1,"created":"2021-08-12 10:40:14.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/78/804378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/804378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/804378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/804378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/804378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/804378/1"}}},"commit":{"parents":[{"commit":"8b2fbe6b64b038e0d0adc4423835b4605e0e4cea","subject":"Merge \"[Docs] Add api-ref for the address groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8b2fbe6b64b038e0d0adc4423835b4605e0e4cea"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-12 10:38:37.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-12 10:39:40.000000000","tz":480},"subject":"Move packet_rate_limit type to neutron-lib","message":"Move packet_rate_limit type to neutron-lib\n\nThis is the TODO work of:\nneutron/services/qos/constants.py\n\nRelated-bug: #1912460\nChange-Id: I295d8004efd563a5044631ec504d4eb9ae1be55d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cb209a112e13e0a8de9605173bac816e08e5a11e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cb209a112e13e0a8de9605173bac816e08e5a11e"}]},"branch":"refs/heads/master"},"0c259e0d8fff12ad492d49e1b47da4bb55416b11":{"kind":"REWORK","_number":2,"created":"2021-08-13 01:32:15.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/78/804378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/804378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/804378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/804378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/804378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/804378/2"}}},"commit":{"parents":[{"commit":"8b2fbe6b64b038e0d0adc4423835b4605e0e4cea","subject":"Merge \"[Docs] Add api-ref for the address groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8b2fbe6b64b038e0d0adc4423835b4605e0e4cea"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-12 10:38:37.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-13 01:24:33.000000000","tz":480},"subject":"Move packet_rate_limit type to neutron-lib","message":"Move packet_rate_limit type to neutron-lib\n\nThese are the TODO works of:\nneutron/services/qos/constants.py\nneutron/extensions/qos_pps_rule.py\n\nRelated-bug: #1912460\nChange-Id: I295d8004efd563a5044631ec504d4eb9ae1be55d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0c259e0d8fff12ad492d49e1b47da4bb55416b11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0c259e0d8fff12ad492d49e1b47da4bb55416b11"}]},"branch":"refs/heads/master"},"8ca7087ed978e412c4b31e543f3c31b824fbb779":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-08-17 05:57:45.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/78/804378/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/804378/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/804378/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/804378/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/804378/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/804378/3"}}},"commit":{"parents":[{"commit":"83bc3878ebed767acf23dbf9cf8ac7e9d9b64c11","subject":"Merge \"Allow PUT on direction param in QoS minimum pps rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/83bc3878ebed767acf23dbf9cf8ac7e9d9b64c11"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-12 10:38:37.000000000","tz":480},"committer":{"name":"liuyulong","email":"i@liuyulong.me","date":"2021-08-17 05:57:45.000000000","tz":0},"subject":"Move packet_rate_limit type to neutron-lib","message":"Move packet_rate_limit type to neutron-lib\n\nThese are the TODO works of:\nneutron/services/qos/constants.py\nneutron/extensions/qos_pps_rule.py\n\nRelated-bug: #1912460\nChange-Id: I295d8004efd563a5044631ec504d4eb9ae1be55d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8ca7087ed978e412c4b31e543f3c31b824fbb779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8ca7087ed978e412c4b31e543f3c31b824fbb779"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
