)]}'
{"id":"openstack%2Fneutron-lib~805010","triplet_id":"openstack%2Fneutron-lib~master~Ic3718defd74c2165077bb5a942e46474d3582cf5","project":"openstack/neutron-lib","branch":"master","topic":"bp/qos-minimum-guaranteed-packet-rate","hashtags":[],"change_id":"Ic3718defd74c2165077bb5a942e46474d3582cf5","subject":"Add util to generate resource request group UUID","status":"MERGED","created":"2021-08-18 11:43:09.000000000","updated":"2021-08-19 15:30:12.000000000","submitted":"2021-08-19 15:27:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"805010-bp/qos-minimum-guaranteed-packet-rate","meta_rev_id":"73575666befd97419116f81d43f8901e3fc77b5f","_number":805010,"virtual_id_number":805010,"owner":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 15:27:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-08-19 09:38:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-08-19 11:05:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2021-08-18 12:25:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-08-19 11:05:46.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-18 11:44:50.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2021-08-18 11:44:50.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-08-18 11:44:50.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-08-18 11:44:50.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-18 12:55:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-18 13:13:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-08-19 09:38:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-08-19 11:05:46.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"46a5f4626898904622f360705d5447a54574f898","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-18 11:43:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b02bcb2686df5440edbf8206b775eb0d5aacdf2c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-18 12:25:01.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nLooks good to me.\nWithin a port different rules end up in different groups (today) so the group id unique as the rule id is unique.\nBetween ports using the same QoS policy the port id will be unique.","accounts_in_message":[],"_revision_number":1},{"id":"7d1cd21e701ba78922b005351bd73db7f55058a3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-18 12:25:52.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Looks good to me.\n\u003e Within a port different rules end up in different groups (today) so the group id unique as the rule id is unique.\n\u003e Between ports using the same QoS policy the port id will be unique.\n\nIgnore this ^^. What I want to say is inline.  (gerrit saved my preliminary notes :/)","accounts_in_message":[],"_revision_number":1},{"id":"db360d012ec1359098372976487214adb46591f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 12:55:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa47ceb323174eedaa0312449a77dd70 : SUCCESS in 6m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40a76efeed014a90b0abe3789c29fbb7 : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/904eb8f11cb54d9f898f78167db4c2b5 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f71501a2b2994da69790eca704d4586d : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2fd9783b5ac4fe0a07bdcecc472b933 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6736efe79fd2439097449fb5d41bd4df : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6dd4965dbd34600a081764faebe75d6 : SUCCESS in 4m 04s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/079d75b3386d4fa7a7307dbf1f75679e : SUCCESS in 1h 10m 56s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/07a6fb28779541f78694019bdca9180d : SUCCESS in 35m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/db34bf8579f8453882ed9ebef40f7589 : FAILURE in 51m 29s","accounts_in_message":[],"_revision_number":1},{"id":"235f45f5dde46898888476e9389d80250ea555b2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-18 13:13:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5075a7efcffb5dc045ec8589f96c2b3f8c5b9ab3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 14:41:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37fb50a225a84ad286a1d85f8249f615 : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf94417691b34f0b90d57c921654edd6 : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99b711fbb2d24ec2ab6d482a3b55b88b : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ca7dc7fdee14013ba15393776c331bf : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4690c0a7c65f4335af7c75d9ba970736 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21ac7f13d8cd448ca3ddf5abed8c8e45 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7679680ca1b54fe2a800559ce80c8721 : SUCCESS in 4m 44s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55b9e4c794714aa49a9d53e5e5594fe6 : SUCCESS in 1h 22m 49s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/94727756bd78492fbd9c64526a5ecc70 : SUCCESS in 41m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a77bff9fb2fa43858e60f80ae50f57b9 : SUCCESS in 47m 15s","accounts_in_message":[],"_revision_number":1},{"id":"cfbcfed03632cc6fd3da4de195c7440b192d0ce8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-19 09:38:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8afc2f997d44d09cfd90944a20e06dd3f927278b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-08-19 11:05:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a17e1a644aa363bcd57d102c910a0b93095037f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 11:06:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"874a34da45318e4337564c75ad7a0494d4c5b7a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 12:32:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c427c513782d45b2a023b94cc978581d : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39babdac58e542d28a58884bd4febe25 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b61b6adcd15049d0bafc72c20ef03ea3 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d525ac19e50a4243b7e91a879d48379f : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54b72620976c44f2b6b08f75426e0ba4 : SUCCESS in 4m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00d3281f79624e929e4127ec7d93ecbe : SUCCESS in 1h 20m 29s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cf52229448e54573affed9a984ee362e : SUCCESS in 44m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd1e808296e24b2c8bc906740d88f3bb : FAILURE in 47m 16s","accounts_in_message":[],"_revision_number":1},{"id":"151209ee0aecbbcbe9c072e8732afa0413d49123","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-19 12:33:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a2a12ab820dfb339099e0f1c2f4a9a00c752cba8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 14:03:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b5065125bc24df3b2905e2e82be120b : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e73afdf8685f414caaf2a85dbfc8b215 : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e1595058dd249fc861a1269d6cbd367 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/661b5a4030fc41b9aa1b98af6d1301f6 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2072ede84ea7436e93f11947398046dd : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fb81baac79f46b4bfd112035780a6c7 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c68c730b68d34c2e985228dd055adc92 : SUCCESS in 4m 45s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cea3094958084ee4becb5af58a4b2208 : SUCCESS in 1h 25m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/57c1a47fb6df4c798aafdc9bc808bd33 : SUCCESS in 48m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ebb6e1d00fd4388883beeb9cfa1a1c2 : SUCCESS in 49m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9fc85e124fc00a12cff83df51ea3dc0ac2acce91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 14:03:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4c536c8429ef5d7aca1fff1798f98a7afd4d781","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:27:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bc75406f0194960940931f3a7caf399 : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ac96c607f7e41d2a38059f7c28b64d9 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2275baa140144d39cf947e1d23588b7 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2b128750baa479fa0808b5b92fb0032 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/205719a5ad1f43aa93c30f2b9ab833fe : SUCCESS in 4m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ba4a1cf0b514962bf5287ab2b14bbd1 : SUCCESS in 1h 19m 18s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/20ff062c969e4627a74fe0eb81a380c3 : SUCCESS in 46m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9ac47a3a01c47538753e99d379fedd1 : SUCCESS in 47m 53s","accounts_in_message":[],"_revision_number":1},{"id":"65bd436ae9214b4ff52aa50163809155f9e1acfb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:27:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"73575666befd97419116f81d43f8901e3fc77b5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:30:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/122cd7bea556493da38b193ebbec5277 : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"794fc7032efca33c24e229d94f914be632911124","revisions":{"794fc7032efca33c24e229d94f914be632911124":{"kind":"REWORK","_number":1,"created":"2021-08-18 11:43:09.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/10/805010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/10/805010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/10/805010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/10/805010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/10/805010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/10/805010/1"}}},"commit":{"parents":[{"commit":"272bb88a86fe4d6eaad4e61a2f92b5bd44a773ee","subject":"Merge \"Move packet_rate_limit type to neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/272bb88a86fe4d6eaad4e61a2f92b5bd44a773ee"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 11:24:06.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 11:32:45.000000000","tz":120},"subject":"Add util to generate resource request group UUID","message":"Add util to generate resource request group UUID\n\nAdd a new utility function that can be used to generate a stable UUID\nfor a resource request group.\n\nRelated-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ic3718defd74c2165077bb5a942e46474d3582cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/794fc7032efca33c24e229d94f914be632911124"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/794fc7032efca33c24e229d94f914be632911124"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
