)]}'
{"id":"openstack%2Fneutron-lib~806102","triplet_id":"openstack%2Fneutron-lib~master~Ida9115cb6d80296921eb04144396bfe6830facb2","project":"openstack/neutron-lib","branch":"master","topic":"dnm","hashtags":[],"change_id":"Ida9115cb6d80296921eb04144396bfe6830facb2","subject":"DNM: test muster","status":"ABANDONED","created":"2021-08-26 08:23:29.000000000","updated":"2021-08-26 10:21:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"14ead9d4d8481346a8d65738c396f0ebfe0b8914","_number":806102,"virtual_id_number":806102,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-26 09:48:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 09:48:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a446cd777df1a594232602dc4617f5ebfb53989","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-26 08:23:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"979daf8afedc0a3d7c2255e3b9f6df2272446e8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 09:48:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83551f9498664313be52dc83baf75710 : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c1d1411ba6f44beb9c6d7961fa9bfcc : FAILURE in 5m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edccbccb9cbd44b79eeb13844cc255fa : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2d9473d87d9440196e034c88dd98cdf : FAILURE in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8d338927f28462a8e8c4e155fd914a5 : FAILURE in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbc7db8473fd4910a09df8f887f81234 : FAILURE in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46dadab9ebcc4cf6a307f285545d17cb : FAILURE in 4m 44s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31dd775d8d544ef7add2f6612bb74ad3 : SUCCESS in 1h 22m 53s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/01f62b4c708d43528b46aa4a06763fcd : SUCCESS in 41m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/947c26d163b7494bab4618e2da08cb93 : FAILURE in 42m 32s","accounts_in_message":[],"_revision_number":1},{"id":"14ead9d4d8481346a8d65738c396f0ebfe0b8914","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-26 10:21:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"295e843eafad4b82c26af38757a95793b2bacaee","revisions":{"295e843eafad4b82c26af38757a95793b2bacaee":{"kind":"REWORK","_number":1,"created":"2021-08-26 08:23:29.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/806102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/02/806102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/02/806102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/02/806102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/02/806102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/02/806102/1"}}},"commit":{"parents":[{"commit":"108a3586c5af9b3b599cab24b917566011688e29","subject":"Merge \"Add network_ip_availability pagging and sorting support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/108a3586c5af9b3b599cab24b917566011688e29"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-26 08:22:56.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-26 08:22:56.000000000","tz":120},"subject":"DNM: test muster","message":"DNM: test muster\n\nChange-Id: Ida9115cb6d80296921eb04144396bfe6830facb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/295e843eafad4b82c26af38757a95793b2bacaee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/295e843eafad4b82c26af38757a95793b2bacaee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
