)]}'
{"id":"openstack%2Fneutron-lib~806129","triplet_id":"openstack%2Fneutron-lib~master~I8611e140305636685c1532dc36812d24a234dc9b","project":"openstack/neutron-lib","branch":"master","topic":"bug/1941725","hashtags":[],"change_id":"I8611e140305636685c1532dc36812d24a234dc9b","subject":"Disable \"oslo.messaging\" metrics","status":"MERGED","created":"2021-08-26 09:59:53.000000000","updated":"2021-08-26 14:16:46.000000000","submitted":"2021-08-26 14:15:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"806129-bug/1941725","meta_rev_id":"165e6fc6870c08019566d0671130b2037c44b09c","_number":806129,"virtual_id_number":806129,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-26 14:15:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-08-26 12:39:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-08-26 11:38:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-08-26 11:25:08.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-08-26 12:39:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 10:12:43.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-08-26 11:25:08.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-08-26 11:38:36.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-08-26 12:29:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 12:39:43.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"6d20c9c5d1e3cdb303dd380c78bcd8a36e77e615","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-26 09:59:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f611341e84e6f00d8ba3e0b17ab0db9bc02f9144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-26 10:05:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9ccf4e126bfc8e1283b414fc9e097657c4451ccf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-26 10:09:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"66e77ff44f9206521efab36d67d6c488b69021cb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-26 10:12:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"81c8b01324bc54cfde77aced4c0d385f89bee836","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-26 10:16:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"90677b69af87078e8e1b983c8bad68d1dd1bc745","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-26 10:20:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"166269fdffa5a8684c1b583c72114ef4e08d9140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-26 10:57:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"950469c11639e83401b88c8d58bb16cf963e89bc","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-08-26 11:25:08.000000000","message":"Patch Set 4: Code-Review+2\n\nLooking at zuul status I see UT are happy with this fix","accounts_in_message":[],"_revision_number":4},{"id":"4d5430c9b060063457921c70d87ae7194b86e3a2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-26 11:38:36.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e2911a477d17f9b55ebf49cd11fa38d03b1f504d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 12:29:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08487c0b2ba74626b69c76a48bdb4383 : SUCCESS in 7m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb961b00d99747edad45cf22d80c8128 : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0700a0b5ac154c15975de4cceb96a9b3 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0664aae8a2814b8999f7260578870817 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15ef3cdc737543c9b649622b38ad3aed : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c648f8f8b8e4189bb097d3a9ef7efc4 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/074603cf1dc34186b3e7178b86c1fc31 : SUCCESS in 4m 28s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9dfb7e368c2d4c338794624b79ac72b8 : SUCCESS in 1h 27m 27s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/987beba2b87549feab3bde3430373e86 : SUCCESS in 42m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8957c1c4670c4067941c6b20019a9614 : SUCCESS in 52m 57s","accounts_in_message":[],"_revision_number":4},{"id":"748f9d1581b2ee3749380e64b95935413544d5c0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-26 12:39:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fb0bdf43cf5cbc4a661dfdb9035e8b4b860ac2d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 12:40:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b65c04098e7fb49ad22edd686bb90b84ab6c4a7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 14:15:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0449d13c08634b278d5e233b2c0ce553 : SUCCESS in 7m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf84e388837a47acacfcfb34a5ab33a5 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28eb6e8a242847b0be32c7ea021620f5 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/843f349e224949ad9dedc7716cd35771 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cefbe1216e804161ad126fd89154912d : SUCCESS in 6m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ced1178f31ed488495d4d626435aeede : SUCCESS in 1h 25m 14s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f8b5e63e80d547719d6cc877a6e5fe5d : SUCCESS in 46m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/45be9d91de2a4b4c953743c3789c0a55 : SUCCESS in 47m 36s","accounts_in_message":[],"_revision_number":4},{"id":"c2e80d23a47e2ca18a706ed3a188db01d6795d61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 14:15:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"165e6fc6870c08019566d0671130b2037c44b09c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 14:16:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/955ebbf02c794cfa9a5bb883c9730a03 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c3f540a4cd252a4147e4e013df114a2303c810b7","revisions":{"7cbfd77acdbf857bdc9aeb2172a26c8bd23c84b3":{"kind":"REWORK","_number":1,"created":"2021-08-26 09:59:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/806129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/806129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/1"}}},"commit":{"parents":[{"commit":"108a3586c5af9b3b599cab24b917566011688e29","subject":"Merge \"Add network_ip_availability pagging and sorting support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/108a3586c5af9b3b599cab24b917566011688e29"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-26 09:54:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-26 09:59:36.000000000","tz":0},"subject":"Provide int value for oslo.messaging \"metrics_buffer_size\"","message":"Provide int value for oslo.messaging \"metrics_buffer_size\"\n\nSince [1], the \"oslo_messaging_metrics.metrics_buffer_size\" should be\na integer value, not a mocked value, to be used with the new\noslo.messaging metrics collection.\n\nThis patch also provides \"metrics_thread_stop_timeout\" value to avoid\nwarning messages during the UT execution.\n\nThe values provided are the default ones.\n\n[1]https://review.opendev.org/c/openstack/oslo.messaging/+/761848\n\nChange-Id: I8611e140305636685c1532dc36812d24a234dc9b\nCloses-Bug: #1941725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7cbfd77acdbf857bdc9aeb2172a26c8bd23c84b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7cbfd77acdbf857bdc9aeb2172a26c8bd23c84b3"}]},"branch":"refs/heads/master"},"3cb5c0dcd62c84bcaa33e248261f1193f53e0d87":{"kind":"REWORK","_number":2,"created":"2021-08-26 10:05:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/806129/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/806129/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/2"}}},"commit":{"parents":[{"commit":"108a3586c5af9b3b599cab24b917566011688e29","subject":"Merge \"Add network_ip_availability pagging and sorting support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/108a3586c5af9b3b599cab24b917566011688e29"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-26 09:54:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-26 10:05:07.000000000","tz":0},"subject":"Provide int value for oslo.messaging \"metrics_buffer_size\"","message":"Provide int value for oslo.messaging \"metrics_buffer_size\"\n\nSince [1], the \"oslo_messaging_metrics.metrics_buffer_size\" should be\na integer value, not a mocked value, to be used with the new\noslo.messaging metrics collection.\n\nThis patch also provides \"metrics_thread_stop_timeout\" value to avoid\nwarning messages during the UT execution.\n\nThe values provided are the default ones.\n\n[1]https://review.opendev.org/c/openstack/oslo.messaging/+/761848\n\nChange-Id: I8611e140305636685c1532dc36812d24a234dc9b\nCloses-Bug: #1941725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3cb5c0dcd62c84bcaa33e248261f1193f53e0d87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3cb5c0dcd62c84bcaa33e248261f1193f53e0d87"}]},"branch":"refs/heads/master"},"46023e36050f37e6091be38c30a6b8c1d469f5df":{"kind":"REWORK","_number":3,"created":"2021-08-26 10:09:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/806129/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/806129/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/3"}}},"commit":{"parents":[{"commit":"108a3586c5af9b3b599cab24b917566011688e29","subject":"Merge \"Add network_ip_availability pagging and sorting support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/108a3586c5af9b3b599cab24b917566011688e29"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-26 09:54:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-26 10:08:08.000000000","tz":0},"subject":"Provide int value for oslo.messaging \"metrics_buffer_size\"","message":"Provide int value for oslo.messaging \"metrics_buffer_size\"\n\nSince [1], the \"oslo_messaging_metrics.metrics_buffer_size\" should be\na integer value, not a mocked value, to be used with the new\noslo.messaging metrics collection.\n\nThe value provided is the default one: 1000.\n\nThis patch also provides \"metrics_thread_stop_timeout\" value to avoid\nwarning messages during the UT execution.\n\nThe values provided is 0, that will prevent from trying to send any\nmetric during the unit tests.\n\n[1]https://review.opendev.org/c/openstack/oslo.messaging/+/761848\n\nChange-Id: I8611e140305636685c1532dc36812d24a234dc9b\nCloses-Bug: #1941725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/46023e36050f37e6091be38c30a6b8c1d469f5df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/46023e36050f37e6091be38c30a6b8c1d469f5df"}]},"branch":"refs/heads/master"},"c3f540a4cd252a4147e4e013df114a2303c810b7":{"kind":"REWORK","_number":4,"created":"2021-08-26 10:57:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/806129/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/806129/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/806129/4"}}},"commit":{"parents":[{"commit":"108a3586c5af9b3b599cab24b917566011688e29","subject":"Merge \"Add network_ip_availability pagging and sorting support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/108a3586c5af9b3b599cab24b917566011688e29"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-26 09:54:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-20 15:03:32.000000000","tz":0},"subject":"Disable \"oslo.messaging\" metrics","message":"Disable \"oslo.messaging\" metrics\n\nSince [1], \"oslo.messaging\" provides a feature to send RPC metrics\nto \"oslo.metrics\". This patch is breaking the CI unit tests jobs.\n\nBecause those metrics are not needed in this CI, this patch disables\nthis feature.\n\n[1]https://review.opendev.org/c/openstack/oslo.messaging/+/761848\n\nChange-Id: I8611e140305636685c1532dc36812d24a234dc9b\nCloses-Bug: #1941725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c3f540a4cd252a4147e4e013df114a2303c810b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c3f540a4cd252a4147e4e013df114a2303c810b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
