)]}'
{"id":"openstack%2Fneutron-lib~807869","triplet_id":"openstack%2Fneutron-lib~master~I4dd0d65f0e7b2c5161f725e542b795e0ba4ba2fe","project":"openstack/neutron-lib","branch":"master","topic":"registry_notify","hashtags":[],"change_id":"I4dd0d65f0e7b2c5161f725e542b795e0ba4ba2fe","subject":"Remove deprecated method registry.notify","status":"MERGED","created":"2021-09-08 13:04:31.000000000","updated":"2021-09-22 19:46:14.000000000","submitted":"2021-09-22 19:44:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"807869-registry_notify","meta_rev_id":"72aa5540897c1e81ff504a667c6f538aae89e25d","_number":807869,"virtual_id_number":807869,"owner":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-22 19:44:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-09-20 09:38:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-09-22 16:07:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-21 15:16:17.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-09-22 16:07:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-08 13:14:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-09-08 15:21:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 15:25:08.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-09-10 11:14:25.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2021-09-15 06:56:40.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-09-15 06:56:40.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"de53cf5e06e45afc5ba76ca87d3ab8aab29507c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-08 13:04:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68bb8714ba2490f06b7f81d2b8645a22906ae8d8","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-08 13:04:44.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"307fadd7cac8f21ead19a90c69d0dfc1aee54f36","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-08 13:14:04.000000000","message":"Patch Set 1:\n\nYou have done a fantastic work here. Thanks a lot!\n\nBecause this is an important change and milestone in the n-lib code, please add a release note.","accounts_in_message":[],"_revision_number":1},{"id":"63e234d9b56b89a3a8a8795d2be547628c8eecbd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-08 13:55:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5f97b56889d0ffe64205a260fbaf7d36a5d8231","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-08 13:58:18.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e You have done a fantastic work here. Thanks a lot!\n\u003e \u003e Thanks Rodolfo!\n\u003e Because this is an important change and milestone in the n-lib code, please add a release note.\n\u003e \u003e Done","accounts_in_message":[],"_revision_number":2},{"id":"ad2ab311d77b40d87adcdb58a6a7c193a11eb4fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 15:21:05.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b65630ddf1154947b9a353e84bf89947 : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afb214a36c5548f1b4a81f962ba02380 : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0145a35133b4f7fafe5959d83c8bbc2 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b336d0de54f4629a0deefecc75eccc9 : FAILURE in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a66c46d12f6848659e3dd095d40db210 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ed2bc04a42248c593cdde3a3fc9bc17 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71c841ae70494d89b4d8c8f7b4a6c7bb : SUCCESS in 5m 33s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96477042640e4bb7a601916023d7f273 : SUCCESS in 1h 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fec0b77cbde40848c4693e4d30aef00 : SUCCESS in 3m 44s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/284bd89c43bd444aabd06789bedef754 : SUCCESS in 1h 05m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e884b7b76e54ef58d83495bb353d921 : FAILURE in 52m 27s","accounts_in_message":[],"_revision_number":2},{"id":"cbb91e200d8615e9b789382b8bbdc1e941880af1","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-09 15:25:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8589e7a528db9ed1f671e4e30bc172e57c205a14","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-09 15:25:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1e655ce9485b195d3b60abe6b1ecfd9be675c1ad","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-09 15:26:30.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"57a3656fc19af077e236f798f2c7690ff7ddccac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 16:52:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/305dcee874bc43359aadc4bb219813cf : SUCCESS in 8m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/459cb5367cb048df94aedda1b40d5108 : SUCCESS in 6m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0319959fb3d472fb9a3f40b74e6f89a : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/928d1cd1002a46a9aca417ddf9ff1e25 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/770a007bbb3d4cccabcdd9971cb10048 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26dedf07215746d89fd9e0bf68403a5f : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4c636a006674157af3261253a455ac6 : SUCCESS in 5m 43s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e533fc102f1c4793ad2e271521871656 : SUCCESS in 1h 24m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfb3efb262514e0e9f932b30e52379a3 : SUCCESS in 5m 08s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/62b2e7148e5f440f92ffd5f0b3c166b5 : SUCCESS in 39m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/281ac6f2da6e49039a6187cb74c896fc : FAILURE in 45m 25s","accounts_in_message":[],"_revision_number":3},{"id":"b4f769f141313f95f7d46ecc67496794245fa39c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-10 11:14:25.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"30b287bf09ea759dff5977f02f4f8aca68d5ea11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-13 07:03:40.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"e20e34c33d7d55e11b3d4b9f125929bb5a94f14d","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-13 07:04:02.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"59d186102f4112308102b7b615a697603c8588ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 08:32:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105f58bca1264b17b5c3bc1b68591e2e : SUCCESS in 7m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f82a83acd8e840c2ba51b20a92d93ac9 : SUCCESS in 6m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2580e2d60a5c4a8585305f2c31fb0691 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ebc06361f324b04beea46e568ca0c89 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f341c7cd7c3942218001aea54b74860d : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef2c43842cac448f8b49818b03ea5f8d : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/971203e15c4641d4b2fa713af679297e : SUCCESS in 6m 21s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59847d17618146629224d8d0aeca4053 : SUCCESS in 1h 22m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f83a3fbdae024c0d8e9fde6caa7074b1 : SUCCESS in 4m 56s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/da30b001c675420f98512b88a7bc603f : SUCCESS in 50m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5f8b9ce46d74936a10e4801d0e2b9b2 : SUCCESS in 47m 32s","accounts_in_message":[],"_revision_number":4},{"id":"dbbf15011c5da7191398566f8e028796a8fbe7ef","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-15 09:26:05.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3c7c996377ed60248b62d8052633b79d5604ed64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-15 09:53:19.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"90efa19edada86b3e7bc267d8fd71b4337b32a0a","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-15 09:53:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1a138972a5f530eea4aec3bd9eb52430253e2d74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-15 09:54:26.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"3b8f6bd244dfc28520172e62825dc2206208042d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 11:31:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/220fe2dcced74daf933b48db5c7fb63d : SUCCESS in 5m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88586d02e3a44ef39fdc5b33d4ec7c9a : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e20f506d5d54cf8aa81ba4ea2599ec9 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf222141a4d42ac8670128c25b4f7aa : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b44e20826c8b431d8871f66c9fb11b68 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22ca3958b1ac46d682dde5996e9193ee : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0a1d0844f6544448b957bb3574a1183 : SUCCESS in 5m 02s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1e5f5d1433c46e184a7985de287f025 : SUCCESS in 1h 22m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bf3b84d5ae647228c353887bc7b3a1b : SUCCESS in 4m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a9c700b4f57f4fb39d00a3b13030f456 : SUCCESS in 42m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/97a98f639c034e03a62501a86ad1a03f : SUCCESS in 46m 19s","accounts_in_message":[],"_revision_number":6},{"id":"c88ff3346a8d8e92c6c32463f42bb3d2c4583373","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-20 09:14:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7937f4e6dd5018f4205fd42bb8efa92d811032b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-20 09:26:21.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"c1b7265e441405da328d8bf9598467df7ca1d44f","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-20 09:28:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"52decc3674b0b7b0f2191d0d9b8ad6d0ecd87432","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-20 09:38:41.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"5c46af3d4f3f90250dd65458b67c63f75808fce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 10:45:42.000000000","message":"Patch Set 7: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aee0ef147dd4fac898f31772b06240c : SUCCESS in 7m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45b838886e1847988d9cc91d51889d4b : SUCCESS in 6m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb866112eb1c42408ed98d6e65228976 : FAILURE in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/775bc9c648c54466b87c890ed3d3edde : FAILURE in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4550b63eb59146d494934498865d9109 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f7e9382b0644a1f9309d20cfc17c1bb : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/874fc20bf61f4d5c8b405fa2eeba6b36 : SUCCESS in 7m 14s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00f0a68e46e14460928e108e20413296 : SUCCESS in 1h 17m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a979484096247ad83ff3ac5120b1137 : SUCCESS in 5m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/13e07f65c70f4e31a3d9ef9fd96e2384 : SUCCESS in 40m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/021a849e71ac4c8bb8003c1c8180c8e6 : SUCCESS in 44m 55s\n\nWarning:\n  Comments left for invalid file neutron_lib/rpc.py\n  Comments left for invalid file neutron_lib/constants.py\n  Comments left for invalid file neutron_lib/api/definitions/address_scope.py\n  Comments left for invalid file neutron_lib/api/definitions/bgp.py\n  Comments left for invalid file neutron_lib/api/definitions/bgpvpn.py\n  Comments left for invalid file neutron_lib/api/validators/dns.py\n  Comments left for invalid file neutron_lib/api/validators/__init__.py\n  Comments left for invalid file neutron_lib/agent/topics.py\n  Comments left for invalid file neutron_lib/exceptions/__init__.py\n  Comments left for invalid file neutron_lib/callbacks/exceptions.py\n  Comments left for invalid file neutron_lib/db/model_base.py\n  Comments left for invalid file neutron_lib/db/api.py\n  Comments left for invalid file neutron_lib/db/utils.py\n  Comments left for invalid file neutron_lib/placement/utils.py\n  Comments left for invalid file neutron_lib/placement/client.py\n  Comments left for invalid file neutron_lib/utils/helpers.py\n  Comments left for invalid file neutron_lib/utils/test.py\n  Comments left for invalid file neutron_lib/utils/net.py\n  Comments left for invalid file neutron_lib/hacking/checks.py","accounts_in_message":[],"_revision_number":7},{"id":"5d6e155bb46da8b9c1f2a089e22b1fa7d610c711","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-21 10:28:21.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"8201f5efa9264a93049f345cbea675a79a17dd49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 11:40:19.000000000","message":"Patch Set 7:\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85858634fcba4352b54146ea66a00492 : SUCCESS in 8m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e09d55b1b3fa433b910ac3c13a226ffa : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/166de21d6bf544b19c82874c8b52187b : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a37514b981e747948c3fe69235213e14 : FAILURE in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cddab35c63984c0bbd4c0fe7d86a57c8 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddeacde5886c4bab9f60c58278900338 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e072d5cd47ef41d5b00f0a1af0379fa4 : SUCCESS in 4m 51s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35efa6c88a0a486ab8d2f58133321c1c : SUCCESS in 1h 08m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29b9d0c6e7a84a28b36d9479ac2a4bdf : SUCCESS in 4m 38s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1de76e25f90d4afc9a331369a31da915 : SUCCESS in 41m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/04b3829088bc4e0190bcd3f5d7469382 : SUCCESS in 46m 02s\n\nWarning:\n  Comments left for invalid file neutron_lib/rpc.py\n  Comments left for invalid file neutron_lib/constants.py\n  Comments left for invalid file neutron_lib/api/definitions/address_scope.py\n  Comments left for invalid file neutron_lib/api/definitions/bgp.py\n  Comments left for invalid file neutron_lib/api/definitions/bgpvpn.py\n  Comments left for invalid file neutron_lib/api/validators/dns.py\n  Comments left for invalid file neutron_lib/api/validators/__init__.py\n  Comments left for invalid file neutron_lib/agent/topics.py\n  Comments left for invalid file neutron_lib/exceptions/__init__.py\n  Comments left for invalid file neutron_lib/callbacks/exceptions.py\n  Comments left for invalid file neutron_lib/db/model_base.py\n  Comments left for invalid file neutron_lib/db/api.py\n  Comments left for invalid file neutron_lib/db/utils.py\n  Comments left for invalid file neutron_lib/placement/utils.py\n  Comments left for invalid file neutron_lib/placement/client.py\n  Comments left for invalid file neutron_lib/utils/helpers.py\n  Comments left for invalid file neutron_lib/utils/test.py\n  Comments left for invalid file neutron_lib/utils/net.py\n  Comments left for invalid file neutron_lib/hacking/checks.py","accounts_in_message":[],"_revision_number":7},{"id":"24bb1160a7d407d3175c8cedbbbb5e16a6941ae9","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-21 15:00:14.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"345384ba0fcb754c80a469270c884dc95579e792","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-21 15:16:17.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"e7a19a2d1b03f10a097b1799b0305d7d9d7d23c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 16:31:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f40ebffa2a664d8892a8fa0fd1d308a8 : SUCCESS in 7m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c69ac0eb7a854e65aba1fce13e95120f : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1517cd6db1bd4c979ccc87ac5a90ba3c : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2f51c8a8ee941ac82aa0f7e83b78483 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28e5ecfb82044596839be4ea50e3a891 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98a63498448f46ee8543528608191207 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f20e9daa289743d7a351cf5cb41da9bc : SUCCESS in 4m 52s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6c6e5e512ff428893d829d564a56094 : SUCCESS in 1h 24m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ce512266cea4171af9fd69585a7a265 : SUCCESS in 5m 14s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f4ea6c2cdd5444d992580e6320f86767 : SUCCESS in 55m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2876a50735b4c58b76eedf61f4605c9 : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":7},{"id":"bd8ebf9261f9992a4b0c5692bffc703c4e297929","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-22 16:07:25.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nbased on the Hound (https://codesearch.opendev.org/?q\u003dnotify%5C(resources\u0026i\u003dnope\u0026literal\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d), we are safe.\nThanks","accounts_in_message":[],"_revision_number":7},{"id":"705b46b77238ef8952658346b23d7a0af0710d85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 17:23:40.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"8bc89e0a40f72d1f3b5eb37fe4a6caa47391d6b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 18:21:41.000000000","message":"Patch Set 7:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"bc3c7dd9c61d9b8ca6a5d3819b86e2dad55de56e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 19:44:46.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eced35d25d5c49e697b98326b87089e7 : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17e06040e20745d39b88ed35356a00eb : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d33d71139dcd4f0e9836b4405538cfd5 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcb5d4594e9e4b67bb579db67cc7bf31 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fe6c165e25c47caaa156cf1589af8ec : SUCCESS in 8m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78bfdc49c1464fdbb21a60ba79179064 : SUCCESS in 1h 18m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87a250cd5a4f4bf4b15015fd3161e479 : SUCCESS in 4m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/13d6a95e574d4faba03875d7b34099de : SUCCESS in 45m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5e9d4a4926d4934b4bef6c02ffdbfcb : SUCCESS in 44m 42s","accounts_in_message":[],"_revision_number":7},{"id":"ed4aee1e1ebaeda526cdcd6784e1791b0685a350","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 19:44:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"72aa5540897c1e81ff504a667c6f538aae89e25d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 19:46:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d970e35bfc4149b45402e977b320a7 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a01e339183c94a5aa4914dc656ef71ea : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ecb63131ecabaaf07e8a7e2b2ccc08f8672975df","revisions":{"b0a1bfdf4018e66234d000874416f2708e4a459b":{"kind":"REWORK","_number":1,"created":"2021-09-08 13:04:31.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/69/807869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/69/807869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/1"}}},"commit":{"parents":[{"commit":"64e4dc29a424249aeee9d0755f20fa400238283a","subject":"Improve \"get_collection_count\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64e4dc29a424249aeee9d0755f20fa400238283a"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-08 13:01:25.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-08 13:01:25.000000000","tz":180},"subject":"(wip) Remove registry notify","message":"(wip) Remove registry notify\n\nAll Neutron code now is using payload style callback.\nAfter stadium projects code will be switch to payload\nstyle callback notify can be removed\n\nChange-Id: I4dd0d65f0e7b2c5161f725e542b795e0ba4ba2fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b0a1bfdf4018e66234d000874416f2708e4a459b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b0a1bfdf4018e66234d000874416f2708e4a459b"}]},"branch":"refs/heads/master"},"2e9309d22cf3a42d820cc03928d745ee116c8242":{"kind":"REWORK","_number":2,"created":"2021-09-08 13:55:39.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/69/807869/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/69/807869/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/2"}}},"commit":{"parents":[{"commit":"64e4dc29a424249aeee9d0755f20fa400238283a","subject":"Improve \"get_collection_count\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64e4dc29a424249aeee9d0755f20fa400238283a"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-08 13:01:25.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-08 13:53:58.000000000","tz":180},"subject":"Remove depricated method registry.notify","message":"Remove depricated method registry.notify\n\nAll Neutron code now is using payload style callback.\nAfter stadium projects code will be switch to payload\nstyle callback notify() can be removed\n\nChange-Id: I4dd0d65f0e7b2c5161f725e542b795e0ba4ba2fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2e9309d22cf3a42d820cc03928d745ee116c8242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2e9309d22cf3a42d820cc03928d745ee116c8242"}]},"branch":"refs/heads/master"},"97b6364a1c09de36d726e72a8e179972a1f899e6":{"kind":"REWORK","_number":3,"created":"2021-09-09 15:25:44.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/69/807869/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/69/807869/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/3"}}},"commit":{"parents":[{"commit":"64e4dc29a424249aeee9d0755f20fa400238283a","subject":"Improve \"get_collection_count\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64e4dc29a424249aeee9d0755f20fa400238283a"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-08 13:01:25.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-09 15:23:33.000000000","tz":180},"subject":"Remove depricated method registry.notify","message":"Remove depricated method registry.notify\n\nAll Neutron code now is using payload style callback.\nAfter stadium projects code will be switch to payload\nstyle callback notify() can be removed\n\nChange-Id: I4dd0d65f0e7b2c5161f725e542b795e0ba4ba2fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/97b6364a1c09de36d726e72a8e179972a1f899e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/97b6364a1c09de36d726e72a8e179972a1f899e6"}]},"branch":"refs/heads/master"},"02f9d84dbefa85194d94306eca931e48a4d7b704":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-09-13 07:03:40.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/69/807869/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/69/807869/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/4"}}},"commit":{"parents":[{"commit":"64e4dc29a424249aeee9d0755f20fa400238283a","subject":"Improve \"get_collection_count\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64e4dc29a424249aeee9d0755f20fa400238283a"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-08 13:01:25.000000000","tz":180},"committer":{"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-13 07:03:40.000000000","tz":0},"subject":"Remove deprecated method registry.notify","message":"Remove deprecated method registry.notify\n\nAll Neutron code now is using payload style callback.\nAfter stadium projects code will be switch to payload\nstyle callback notify() can be removed\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-dynamic-routing/+/808024\nDepends-On: https://review.opendev.org/c/openstack/networking-odl/+/808028\nChange-Id: I4dd0d65f0e7b2c5161f725e542b795e0ba4ba2fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/02f9d84dbefa85194d94306eca931e48a4d7b704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/02f9d84dbefa85194d94306eca931e48a4d7b704"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e970b5b0bf93cf224938d722400b9a1138be3ccd":{"kind":"REWORK","_number":5,"created":"2021-09-15 09:53:19.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/69/807869/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/69/807869/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/5"}}},"commit":{"parents":[{"commit":"64e4dc29a424249aeee9d0755f20fa400238283a","subject":"Improve \"get_collection_count\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64e4dc29a424249aeee9d0755f20fa400238283a"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-08 13:01:25.000000000","tz":180},"committer":{"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-15 09:53:14.000000000","tz":0},"subject":"Remove deprecated method registry.notify","message":"Remove deprecated method registry.notify\n\nAll Neutron code now is using payload style callback.\nAfter stadium projects code will be switch to payload\nstyle callback notify() can be removed\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-dynamic-routing/+/808024\nDepends-On: https://review.opendev.org/c/openstack/networking-odl/+/808028\nChange-Id: I4dd0d65f0e7b2c5161f725e542b795e0ba4ba2fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e970b5b0bf93cf224938d722400b9a1138be3ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e970b5b0bf93cf224938d722400b9a1138be3ccd"}]},"branch":"refs/heads/master"},"4e6e288fbefdd76d21fc5db617df507a4dd7bd8f":{"kind":"REWORK","_number":6,"created":"2021-09-15 09:54:26.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/69/807869/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/69/807869/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/6"}}},"commit":{"parents":[{"commit":"64e4dc29a424249aeee9d0755f20fa400238283a","subject":"Improve \"get_collection_count\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64e4dc29a424249aeee9d0755f20fa400238283a"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-08 13:01:25.000000000","tz":180},"committer":{"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-15 09:54:22.000000000","tz":0},"subject":"Remove deprecated method registry.notify","message":"Remove deprecated method registry.notify\n\nAll Neutron code now is using payload style callback.\nAfter stadium projects code will be switch to payload\nstyle callback notify() can be removed\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-dynamic-routing/+/808024\nDepends-On: https://review.opendev.org/c/openstack/networking-odl/+/808028\nChange-Id: I4dd0d65f0e7b2c5161f725e542b795e0ba4ba2fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4e6e288fbefdd76d21fc5db617df507a4dd7bd8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4e6e288fbefdd76d21fc5db617df507a4dd7bd8f"}]},"branch":"refs/heads/master"},"ecb63131ecabaaf07e8a7e2b2ccc08f8672975df":{"kind":"REWORK","_number":7,"created":"2021-09-20 09:26:21.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/69/807869/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/69/807869/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/69/807869/7"}}},"commit":{"parents":[{"commit":"64e4dc29a424249aeee9d0755f20fa400238283a","subject":"Improve \"get_collection_count\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64e4dc29a424249aeee9d0755f20fa400238283a"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-08 13:01:25.000000000","tz":180},"committer":{"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-20 09:26:07.000000000","tz":0},"subject":"Remove deprecated method registry.notify","message":"Remove deprecated method registry.notify\n\nAll Neutron code now is using payload style callback.\nAfter stadium projects code will be switch to payload\nstyle callback notify() can be removed\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-dynamic-routing/+/808024\nDepends-On: https://review.opendev.org/c/openstack/networking-odl/+/808028\nChange-Id: I4dd0d65f0e7b2c5161f725e542b795e0ba4ba2fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ecb63131ecabaaf07e8a7e2b2ccc08f8672975df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ecb63131ecabaaf07e8a7e2b2ccc08f8672975df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
