)]}'
{"id":"openstack%2Fneutron-lib~807876","triplet_id":"openstack%2Fneutron-lib~master~I94baf069624da4e4009eb8c0820f91b4d9e6738f","project":"openstack/neutron-lib","branch":"master","topic":"bug/1936408","hashtags":[],"change_id":"I94baf069624da4e4009eb8c0820f91b4d9e6738f","subject":"Add API shim extension \"quota-check-limit\"","status":"MERGED","created":"2021-09-08 13:43:47.000000000","updated":"2021-09-10 15:10:28.000000000","submitted":"2021-09-10 14:44:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"807876-bug/1936408","meta_rev_id":"7166d6a9bce75ecfcf9f15b85896d4df54cb2a71","_number":807876,"virtual_id_number":807876,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 14:44:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-09-10 13:02:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-09-10 13:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-10 07:38:18.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":1,"date":"2021-09-10 07:47:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-09-10 13:02:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-08 14:31:04.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-09-08 15:36:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 07:39:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-09-09 09:27:27.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-09-10 07:47:50.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"12e183c389787d65442d445ea67ee197c5ea27c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-08 13:43:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e51e0950c49ea8878708726c60518be41ea92983","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-08 14:31:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7f7b6c932caa6b3e128a2dbb291eddcf269b3e78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 15:36:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86d2b105a4145868241c0457da2f63e : SUCCESS in 5m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f59caa90eef47ecbf9b1c53d94660d6 : FAILURE in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6de304a266c74baf91e110931fb1441a : FAILURE in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bbc3e869e98491fbd1aaaa83848c89e : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d8a7ef1b6664f4ba9afc27dd0512fbc : FAILURE in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8071bd4b7cc4be29e9cbe2e0c78781c : FAILURE in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6fba2cd161941768ad88ecf9edf0923 : FAILURE in 5m 56s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfdd161845674862ad5f8cff69b5697d : SUCCESS in 1h 40m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1892df970ea1446da03cf4c92ab45446 : SUCCESS in 4m 35s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d1643dad303042bf993042271b5b4b7f : SUCCESS in 42m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/82ab72a03f994874adf27ea6683dedf3 : SUCCESS in 46m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7340e4d67931803bf4ced46f9f44d62722cff3d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-08 15:43:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"420c81e60f8575d6335cc4d7f1ed5194a95d243c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 17:09:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f16be245e24a4c09b3453fb4f20762aa : SUCCESS in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a214064bc4b4962b1c501ef18920934 : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ab48f4cf0f14e13a7595765c973abd5 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af1430539ac4ad7b5910f2c6c2186e6 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42755ab2c91247ff8526d6d0d32d2a99 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5975cf4110864b05ae8a23769fe3f654 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/397dae3ae2fa440b82a679f4d505dfbb : SUCCESS in 5m 32s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d735b517eb894eb6a4425ad425586ede : SUCCESS in 1h 24m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad063b033ac64ac4b51197f1bdd79435 : SUCCESS in 4m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9ff2ee68810f4d0e94b1c94fe75c0608 : SUCCESS in 42m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ee3a83cd22c48aaa06fb5b33074f016 : SUCCESS in 46m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ec7d5ab528d1705ec26a7772e5318296b0682e9a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-09 07:11:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0a44a78e6f0ccdcb11468c566c6ab001b87ca1f1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-09 09:27:27.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nIt should be mentioned in the api-ref as well","accounts_in_message":[],"_revision_number":2},{"id":"f21a3072423722a432d28bb6574804b5bf2ec0ca","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-09 14:03:57.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e It should be mentioned in the api-ref as well\n\nDone","accounts_in_message":[],"_revision_number":2},{"id":"206945246f3942d639213c503dd952460ac78f03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-09 14:33:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"374e134fed199b966ea765194db816d6c94d8b43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 15:57:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4bace347d779433e987cdac3dfb9cb98 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f98cd0ae696c416c8ba04246d02b5cfd : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26c45f40c2654dceb0b9ce0121433b02 : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e427797567224e049244d9646c80cc8f : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89b11ada16a248659f8ff0c061bfaa61 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cde965bb600418582992822bd9d9b05 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e460d78cc14a4555852f3e2e729cd781 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54a24b74552d45688644f7af071b94d7 : SUCCESS in 6m 23s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a86bdaaa6651483c809c748f429cc125 : SUCCESS in 1h 18m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/587dd812f44b4588a711bfff5e7beb2e : SUCCESS in 4m 54s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/73caa0aed61a4440a1713d71180c5ce8 : SUCCESS in 45m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e154e4623bfb4c1395c1c1413b1eb37a : SUCCESS in 50m 21s","accounts_in_message":[],"_revision_number":3},{"id":"f78bf3f2e84d4ef6ca21d8fe69ab8f3973148972","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-10 07:38:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dcbb9be4651e1e1651e14b957e4c42261fededdf","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-09-10 07:47:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"71c4b596b3bb8aefe520ec440cac7cf50dd872b1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-10 07:53:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dbfac57d59c580aaa9c5b62a6c0cbc734a24896d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-10 11:06:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"599765e411cdf509c49ccf4db004e686c62b2a21","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-10 11:19:32.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"acded6ac53442ce21bbd392971d9b74422d216a7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-10 11:43:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1218a2e2727eb5b359fc6bc8f45251770bd9d38d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-10 13:02:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7955bfe739d9812b443899c0bdf041c880a40ca4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 13:03:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"855f2a519a39c07ea4dc7e817e6102ad46a5cd42","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-10 13:37:08.000000000","message":"Patch Set 3: Code-Review+2\n\nI checked, and it works as expected.","accounts_in_message":[],"_revision_number":3},{"id":"e9ed3f4c2da1d61709a46e79b8d455d92180829a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 14:44:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ea6dfc144d194218a48e0cacba5696d7 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/319660ada10e4db6a9e4a40e28fa2f94 : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b068caf573d2439b97b47913ed593672 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5b4034b6fcd4f8292db7c55319dbd00 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ac6a127249047df9812719b1d9306f9 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76dd1aaee5144e2c9e85d15472b16ef8 : SUCCESS in 4m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16fa35131c17492f992c84cc786a4aa1 : SUCCESS in 1h 27m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a983e1e708848d7bef21896ce7cd0e8 : SUCCESS in 4m 40s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6326e9f4b5da482494590ac5cc6788cc : SUCCESS in 44m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8d9bb6a091b4b3c8f3a3d27a9420ef5 : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":3},{"id":"3d61bbd0d6bf62b9d4bbee214f68807846a1d7f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 14:44:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7166d6a9bce75ecfcf9f15b85896d4df54cb2a71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:10:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/40420abe8e474edf969c5a7936cca05b : SUCCESS in 2m 17s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f8ff91d33824f2d8c9e155c3d0b8cb6 : SUCCESS in 2m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18b2809b29694d2487d53b8613c8b589 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"14df80519c111a896f57886dc5db93284682dbba","revisions":{"132395d1a506c85e5658968dcc6f7bdd651825c3":{"kind":"REWORK","_number":1,"created":"2021-09-08 13:43:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/807876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/76/807876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/807876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/807876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/807876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/76/807876/1"}}},"commit":{"parents":[{"commit":"64e4dc29a424249aeee9d0755f20fa400238283a","subject":"Improve \"get_collection_count\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64e4dc29a424249aeee9d0755f20fa400238283a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-09-08 13:42:43.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-09-08 13:42:43.000000000","tz":0},"subject":"Add API shim extension \"quota-check-limit\"","message":"Add API shim extension \"quota-check-limit\"\n\nThis extension indicates if Neutron supports checking the resource\nusage before applying a new quota limit.\n\nChange-Id: I94baf069624da4e4009eb8c0820f91b4d9e6738f\nRelated-Bug: #1936408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/132395d1a506c85e5658968dcc6f7bdd651825c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/132395d1a506c85e5658968dcc6f7bdd651825c3"}]},"branch":"refs/heads/master"},"55790655d5eccc5bf22f572c414fa3ec0779af5a":{"kind":"REWORK","_number":2,"created":"2021-09-08 15:43:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/807876/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/76/807876/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/807876/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/807876/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/807876/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/76/807876/2"}}},"commit":{"parents":[{"commit":"64e4dc29a424249aeee9d0755f20fa400238283a","subject":"Improve \"get_collection_count\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64e4dc29a424249aeee9d0755f20fa400238283a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-09-08 13:42:43.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-09-08 15:43:06.000000000","tz":0},"subject":"Add API shim extension \"quota-check-limit\"","message":"Add API shim extension \"quota-check-limit\"\n\nThis extension indicates if Neutron supports checking the resource\nusage before applying a new quota limit.\n\nChange-Id: I94baf069624da4e4009eb8c0820f91b4d9e6738f\nRelated-Bug: #1936408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/55790655d5eccc5bf22f572c414fa3ec0779af5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/55790655d5eccc5bf22f572c414fa3ec0779af5a"}]},"branch":"refs/heads/master"},"14df80519c111a896f57886dc5db93284682dbba":{"kind":"REWORK","_number":3,"created":"2021-09-09 14:33:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/807876/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/76/807876/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/807876/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/807876/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/807876/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/76/807876/3"}}},"commit":{"parents":[{"commit":"64e4dc29a424249aeee9d0755f20fa400238283a","subject":"Improve \"get_collection_count\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64e4dc29a424249aeee9d0755f20fa400238283a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-09-08 13:42:43.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-09-09 14:33:28.000000000","tz":0},"subject":"Add API shim extension \"quota-check-limit\"","message":"Add API shim extension \"quota-check-limit\"\n\nThis extension indicates if Neutron supports checking the resource\nusage before applying a new quota limit.\n\nChange-Id: I94baf069624da4e4009eb8c0820f91b4d9e6738f\nRelated-Bug: #1936408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/14df80519c111a896f57886dc5db93284682dbba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/14df80519c111a896f57886dc5db93284682dbba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
