)]}'
{"id":"openstack%2Fneutron-lib~809978","triplet_id":"openstack%2Fneutron-lib~master~I0505a7d5f42c972ff68acbef1ab7794600824e76","project":"openstack/neutron-lib","branch":"master","topic":"bug/1907089","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-10-13 08:25:40.000000000","reason":"Rodolfo Alonso replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-10-18 07:37:48.000000000","reason":"Slawek Kaplonski replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2021-10-18 09:00:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0505a7d5f42c972ff68acbef1ab7794600824e76","subject":"BFD validator","status":"MERGED","created":"2021-09-20 09:07:02.000000000","updated":"2024-03-26 14:30:45.000000000","submitted":"2021-10-18 09:00:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"809978-bug/1907089","meta_rev_id":"d3bbdeefd6e8d611ad4350376f9b8191994e7e0a","_number":809978,"virtual_id_number":809978,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-18 09:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-10-14 08:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-10-18 07:37:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-10-18 07:37:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}],"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-20 10:28:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-11 10:07:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-10-13 09:55:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-03-26 14:30:45.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"}],"messages":[{"id":"d28d2748dd6c06bd8f8880455b4478da918e0b83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-20 09:07:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be74021f93452c15ba2daa638b3c7476048d6433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 10:28:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b897c89e25c34db4a0af09d0c5887e6b : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0fa3285985048fa9d481a3087f4ade8 : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef0496e8a7b5426d8f204beedd4aa509 : FAILURE in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4edb70b719e44170b7b4636f12a4defd : FAILURE in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46fde799d250466aba82a7933ecf33c9 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c651ab354b94be8800eba012ffb9d70 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98ed949ca4ab47aa9045a4b3ad6d7a3f : SUCCESS in 5m 20s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48dbf49455034225805cff8e8aad4682 : SUCCESS in 1h 19m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9457ac00bd914a4a823b6d975c065b68 : SUCCESS in 40m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/91bb81eb6fb34eea9f4ec092eddeab00 : SUCCESS in 41m 32s\n\nWarning:\n  Comments left for invalid file neutron_lib/rpc.py\n  Comments left for invalid file neutron_lib/worker.py\n  Comments left for invalid file neutron_lib/constants.py\n  Comments left for invalid file neutron_lib/api/definitions/address_scope.py\n  Comments left for invalid file neutron_lib/api/definitions/bgp.py\n  Comments left for invalid file neutron_lib/api/definitions/bgpvpn.py\n  Comments left for invalid file neutron_lib/api/validators/dns.py\n  Comments left for invalid file neutron_lib/api/validators/__init__.py\n  Comments left for invalid file neutron_lib/agent/topics.py\n  Comments left for invalid file neutron_lib/exceptions/__init__.py\n  Comments left for invalid file neutron_lib/callbacks/manager.py\n  Comments left for invalid file neutron_lib/callbacks/exceptions.py\n  Comments left for invalid file neutron_lib/db/model_base.py\n  Comments left for invalid file neutron_lib/db/api.py\n  Comments left for invalid file neutron_lib/db/utils.py\n  Comments left for invalid file neutron_lib/placement/utils.py\n  Comments left for invalid file neutron_lib/placement/client.py\n  Comments left for invalid file neutron_lib/utils/helpers.py\n  Comments left for invalid file neutron_lib/utils/test.py\n  Comments left for invalid file neutron_lib/utils/net.py\n  Comments left for invalid file neutron_lib/hacking/checks.py","accounts_in_message":[],"_revision_number":1},{"id":"08e774bb412bdb735a31038b86138e533a5c5b46","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-11 10:07:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"61fb1079d33040596d76aa34f77d33b0e75ec318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-12 09:49:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"564b41eda770cb827b043ae682423d576c374927","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-12 09:50:38.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ee6508b7f324d1c531975ce6202d1faa3df35720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 11:32:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af982814851d45dba4257f9b036b25e8 : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fdbcc91de15409db3f02eb4840b4b0c : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03a38e333c0e412c9706d0d96e6ddbba : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b4e205e530e49399903113d3836e293 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3564110f74a3401fa1ec07c828003e0b : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1627fd9654294f68bf8b27433134f554 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d7ec5bce01140648899ee4f4be0b2d9 : SUCCESS in 4m 45s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e69e45229e143acb2f430913b9e7bb4 : SUCCESS in 1h 36m 20s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e77d6bda06e044c0a6b233645ad12878 : SUCCESS in 38m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6393d2dc079d4a61bf7e3140abf4df7e : SUCCESS in 49m 38s","accounts_in_message":[],"_revision_number":2},{"id":"50535a9f7b5ad64b9a82e924b9ba10919a647064","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-13 08:25:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3a99846f1e9227eeb7f59ab283569748d75c1c93","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-13 09:55:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"156a50bb3d24cc91c27f0d230ad17a423a9b4474","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-14 07:35:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dc6066871343ccadb54acab7e18b80cd3e5840b3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-14 07:53:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2cf4bdd635d9e31fb4a98047156fafb8f7a33136","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-14 08:20:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e1bf969003aa664887cefba5419ec1fc417ca2ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 12:08:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce5f738e94164a6682c8afaa1e160814 : SUCCESS in 8m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91c88e5135764f0e8cb08ad732b9dbd8 : SUCCESS in 7m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b601d13dcbda46d5abae1ef11905b8ee : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff58421bc2a44f09a3645815c80e525a : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4242c9de30ee471aad4cf8a0536b839e : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92123d5bf4ba4d799eeb9021105271fb : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dd06bcbc6bc48ffa5d8bc2317268e8d : SUCCESS in 6m 59s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2b40fe9de8146cb8d59fd6317862822 : SUCCESS in 1h 31m 47s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/053508634cf442938493fac7cd129f86 : SUCCESS in 1h 02m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/88dcf57123c84b6298a72b588b3c456f : SUCCESS in 49m 00s","accounts_in_message":[],"_revision_number":3},{"id":"ff326ff38ae598e3e38c36d63f06bf89dfc72c10","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-18 07:37:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2a0b080b20a4394696f1e361bdf2fb04a973f329","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 07:38:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"482c633d6a8657e4362e3b9ced912dfcfcf9e3c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 09:00:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9345f984c91547c784ec990b2c1d4e9f : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8848dd59edc4dcfb932bea2973dc033 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d49e725bb143b3adbabd3712df7a42 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/799cce8e71854678aa064479d113878e : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b3da89b26d94e82b33845f08296a902 : SUCCESS in 4m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bda56da872454a91bdadaa00f4fed455 : SUCCESS in 1h 18m 35s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d86f498d205445449a5410b58b4d62db : SUCCESS in 52m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7480a6fb23ff4e63b96458440a77ab3b : SUCCESS in 47m 10s","accounts_in_message":[],"_revision_number":3},{"id":"3d6d22a85b3037efeaab1450a296281cbb6abaa1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 09:00:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b858b68d9ab71db6a94feb5c81c0cd99abfee5f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 09:02:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d1ac2ae1c04d6ab87f64e1007a2693 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d126257982fb9b457493115b55905d688ae9a499","revisions":{"a6aa699c2ba821490e4e84d7c97d0b5c6468df36":{"kind":"REWORK","_number":1,"created":"2021-09-20 09:07:02.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/78/809978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/809978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/809978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/809978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/809978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/809978/1"}}},"commit":{"parents":[{"commit":"890d62a3df3f35bb18bf1a11e79a9e97e7dd2d2c","subject":"Merge \"Add Python3 yoga unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/890d62a3df3f35bb18bf1a11e79a9e97e7dd2d2c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-17 08:40:33.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-20 09:05:39.000000000","tz":120},"subject":"BFD validator","message":"BFD validator\n\nAdd mode validator and authenticat type validator for BFD.\n\nChange-Id: I0505a7d5f42c972ff68acbef1ab7794600824e76\nRelated-Bug: #1907089\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a6aa699c2ba821490e4e84d7c97d0b5c6468df36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a6aa699c2ba821490e4e84d7c97d0b5c6468df36"}]},"branch":"refs/heads/master"},"c29b47822750e421ff683f0bc889e6a64a1d1986":{"kind":"REWORK","_number":2,"created":"2021-10-12 09:49:24.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/78/809978/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/809978/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/809978/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/809978/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/809978/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/809978/2"}}},"commit":{"parents":[{"commit":"890d62a3df3f35bb18bf1a11e79a9e97e7dd2d2c","subject":"Merge \"Add Python3 yoga unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/890d62a3df3f35bb18bf1a11e79a9e97e7dd2d2c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-17 08:40:33.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-10-12 09:49:05.000000000","tz":120},"subject":"BFD validator","message":"BFD validator\n\nAdd mode validator and authenticat type validator for BFD.\n\nChange-Id: I0505a7d5f42c972ff68acbef1ab7794600824e76\nRelated-Bug: #1907089\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c29b47822750e421ff683f0bc889e6a64a1d1986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c29b47822750e421ff683f0bc889e6a64a1d1986"}]},"branch":"refs/heads/master"},"d126257982fb9b457493115b55905d688ae9a499":{"kind":"REWORK","_number":3,"created":"2021-10-14 07:35:41.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/78/809978/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/809978/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/809978/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/809978/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/809978/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/809978/3"}}},"commit":{"parents":[{"commit":"890d62a3df3f35bb18bf1a11e79a9e97e7dd2d2c","subject":"Merge \"Add Python3 yoga unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/890d62a3df3f35bb18bf1a11e79a9e97e7dd2d2c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-17 08:40:33.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-10-14 07:35:17.000000000","tz":120},"subject":"BFD validator","message":"BFD validator\n\nAdd mode validator and authenticat type validator for BFD.\n\nChange-Id: I0505a7d5f42c972ff68acbef1ab7794600824e76\nRelated-Bug: #1907089\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d126257982fb9b457493115b55905d688ae9a499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d126257982fb9b457493115b55905d688ae9a499"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
