)]}'
{"id":"openstack%2Fneutron-lib~810195","triplet_id":"openstack%2Fneutron-lib~stable%2Fxena~Iccda2830b4ef2d512709b3775ffab415ce47581e","project":"openstack/neutron-lib","branch":"stable/xena","topic":"DNM-stable/xena","hashtags":[],"change_id":"Iccda2830b4ef2d512709b3775ffab415ce47581e","subject":"Bump requirements and disable f-string check","status":"ABANDONED","created":"2021-09-21 15:08:19.000000000","updated":"2021-09-24 15:43:07.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"cherry_pick_of_change":810003,"cherry_pick_of_patch_set":4,"meta_rev_id":"a3ca208b20c4047e7ced7bfd3fd9d7e51a13145d","_number":810195,"virtual_id_number":810195,"owner":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-21 15:34:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"disliked":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-09-21 15:44:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2021-09-24 14:57:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":-1,"date":"2021-09-24 12:53:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2021-09-24 13:13:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-21 15:08:19.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-09-21 15:22:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-21 15:44:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-09-22 13:33:49.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-09-24 12:53:45.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"5bba85c2d93b9abfe5f52c439aab032009daf01a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-21 15:08:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3a9647ae55364b480329cfc31d91842393e262db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 15:22:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8fb12b7364d42aca80b4a83bf5efee5 : SUCCESS in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c93c1b21202448128bea0b556abcd2ad : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72d02bbe551b4fc189291f38eecceab0 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8cc850bd2b642b6b3668902fca5c671 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34b1b00dc40440e2b67c10c23e30eb32 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/514b20cb751043e8ae42ac6d509a4c11 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8033f26bbd934f548a6a56bce5215c8c : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b692adfcdc013513fda9d62f9ad9ca2fe3520350","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-21 15:24:18.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b9d4b9b4b268ed41d95733cbe24296aaeb01459c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 15:34:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee256f9ec517484a8ae640bd87a667e9 : SUCCESS in 7m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eeb02fdfaeca413cba57aefeb2f49aac : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da9006db7bbc4743968295b251cc6803 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffe6007546004f1abf46f0ed99a4685e : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22529cf451ff461285f147bda811e1e9 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcd908785799485bbf77bb01b7e37ea4 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3a60b9ac7c1403b9047879120054fd2 : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"abb359f86782349f58f16d7fffcae9d66a75c0ef","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-21 15:44:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5b1ae745889073c22227fb01eac4a7328ae254f0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-22 15:26:03.000000000","message":"Patch Set 2: Code-Review-1\n\nNot sure if bumping requirements on stable branches is a good approach, see this thread:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-September/024987.html\n\nI am not totally against it, but we have to understand all aspects of it.","accounts_in_message":[],"_revision_number":2},{"id":"28268190f587a51cbed06eabc44e6b6aadefb020","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-24 09:26:17.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e Not sure if bumping requirements on stable branches is a good approach, see this thread:\n\u003e http://lists.openstack.org/pipermail/openstack-discuss/2021-September/024987.html\n\u003e \n\u003e I am not totally against it, but we have to understand all aspects of it.\n\nBackported to stable Xena because on thread was \u0027On stable/xena we can still solve it the same way with  a new RC. But on older stable branches such solution might be against \nthe stable policy as it would require a major bump on our dependencies.\u0027","accounts_in_message":[],"_revision_number":2},{"id":"63a964897f525740cc4c8ddf6280ff2c1622f142","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-24 11:31:44.000000000","message":"Patch Set 2: Code-Review+1\n\nok, I have to ask around the release team if we need to re-release anything","accounts_in_message":[],"_revision_number":2},{"id":"b30ca95e1dcecce382dcb1a9a345d19acc3fed58","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-24 12:53:45.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbad2f57920d90ef002727eb09eed0ede80546db","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-24 13:13:08.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"998c22151a7a00e8eaa96f117e827459c2e155b4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-24 14:57:53.000000000","message":"Patch Set 2: Code-Review-1\n\nOK I added that consider-using-f-string to https://review.opendev.org/c/openstack/neutron-lib/+/808599 (that drops l-c), let\u0027s see there and drop that one if it works","accounts_in_message":[],"_revision_number":2},{"id":"a3ca208b20c4047e7ced7bfd3fd9d7e51a13145d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-24 15:43:07.000000000","message":"Abandoned\n\nFor stable branches decided drop lower constraints: https://review.opendev.org/c/openstack/neutron-lib/+/808599","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5ff59533363f5ca5a05810fc7e2237e0caa43444","revisions":{"afd10070a579f649c6b25523f7c012732599fcc2":{"kind":"REWORK","_number":1,"created":"2021-09-21 15:08:19.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/95/810195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/95/810195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/95/810195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/95/810195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/95/810195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/95/810195/1"}}},"commit":{"parents":[{"commit":"11d1cbf32435698a8a00b0624a71616445d404d9","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/11d1cbf32435698a8a00b0624a71616445d404d9"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-20 12:04:56.000000000","tz":120},"committer":{"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-21 15:08:19.000000000","tz":0},"subject":"[DNM] Bump requirements and disable f-string check","message":"[DNM] Bump requirements and disable f-string check\n\nBump Mako to 1.0.7 in lower-constraints.txt and add consider-using-f-string\nto .pylintrc to avoid pep8 failure at pylint stage.\n\nChange-Id: Iccda2830b4ef2d512709b3775ffab415ce47581e\n(cherry picked from commit c266a3e6d2e45cf783a0902c7c5682602168ea44)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/afd10070a579f649c6b25523f7c012732599fcc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/afd10070a579f649c6b25523f7c012732599fcc2"}]},"branch":"refs/heads/stable/xena"},"5ff59533363f5ca5a05810fc7e2237e0caa43444":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-09-21 15:24:18.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/95/810195/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/95/810195/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/95/810195/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/95/810195/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/95/810195/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/95/810195/2"}}},"commit":{"parents":[{"commit":"11d1cbf32435698a8a00b0624a71616445d404d9","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/11d1cbf32435698a8a00b0624a71616445d404d9"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-20 12:04:56.000000000","tz":120},"committer":{"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-21 15:24:18.000000000","tz":0},"subject":"Bump requirements and disable f-string check","message":"Bump requirements and disable f-string check\n\nBump Mako to 1.0.7 in lower-constraints.txt and add consider-using-f-string\nto .pylintrc to avoid pep8 failure at pylint stage.\n\nChange-Id: Iccda2830b4ef2d512709b3775ffab415ce47581e\n(cherry picked from commit c266a3e6d2e45cf783a0902c7c5682602168ea44)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5ff59533363f5ca5a05810fc7e2237e0caa43444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5ff59533363f5ca5a05810fc7e2237e0caa43444"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
