)]}'
{"id":"openstack%2Fneutron-lib~813557","triplet_id":"openstack%2Fneutron-lib~master~I3478950acb99f686a5ae3758729862b48cc9bd7f","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2021-10-27 15:10:49.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-10-25 10:37:14.000000000","reason":"Slawek Kaplonski replied on the change"},"33273":{"account":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"last_update":"2021-10-27 15:10:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3478950acb99f686a5ae3758729862b48cc9bd7f","subject":"Addition of status and name field to bgp_associations API","status":"MERGED","created":"2021-10-12 05:10:05.000000000","updated":"2021-10-27 15:12:45.000000000","submitted":"2021-10-27 15:10:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813557","meta_rev_id":"fb7ba34d925b7d2920a9fa1754dbd8d9103d5c38","_number":813557,"virtual_id_number":813557,"owner":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-27 15:10:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":2,"date":"2021-10-25 10:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-27 13:51:51.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-27 13:51:51.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-12 06:39:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-12 10:47:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-10-19 10:33:11.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2021-10-27 13:51:51.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"c5ea156e43c0c7037d1017fccc9e9d2990002cf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"date":"2021-10-12 05:10:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7676efb3b39163e5588a9b9d68ddeacb40f17322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 06:39:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c4489977c1d47d8871139e7423cd3f5 : SUCCESS in 5m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9050252e96a14a1a81912f8679804160 : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e32232d0d3d4e0fac5d36fcf1f4202b : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50560c60a98f4ad9a456035283db6c15 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07b81a6708a9471fb1a5a272358f9b9f : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cabb4e6b34442ec9169865f03623d1c : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7877de83e6b94e248ccb37d1e94424c8 : SUCCESS in 5m 10s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/725f8064053b4ea9b6b424d70127e1c1 : SUCCESS in 1h 27m 49s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fec3555cb26f4cfdba687686f5a1106a : SUCCESS in 35m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7da8b09823dd4d87bbf9ad3dc3246a50 : SUCCESS in 41m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5ce785eb55a9a75b2b163262a1ed833909093546","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-12 10:47:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82a2f84876b3222d1243f0e38ed8f56683d09acd","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-10-19 10:33:11.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"362f0017ff338c8560182b90ac1f595e245afd6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"date":"2021-10-25 09:13:35.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"38ca62b680be29f0892b931d21afecabf67c49ce","author":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"date":"2021-10-25 09:40:15.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c6416f4a8b30da12eb6dcbcf9f17de898d8b7123","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 10:29:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/008b7e85e2554f538b778727ecdba7f7 : SUCCESS in 7m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93a305a396c44689adc71f984794de66 : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56bd140610244bcd9f2ffa0b4bc349c9 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb21b32a210746c9a80b9b0cad5dfe2a : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2869c30498a4f82921bd92cd734e5f3 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/989163b6a6fe49a387be50fcd2a92c04 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/337e2022abb54ab7a84d60eb5f8b3d41 : SUCCESS in 4m 41s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8376c0cbe9041c1a815407c2b2688eb : SUCCESS in 1h 14m 04s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/eb7139e74e9044c79d26180b4badb328 : SUCCESS in 43m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/073cc50e3fd642a78b8f38f0738b39d2 : SUCCESS in 47m 00s","accounts_in_message":[],"_revision_number":2},{"id":"e64c1d9b77c1a1a040df573d536390ea3e08064f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-25 10:37:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21cc44f0226986ce77e2f6a463a0211a1310c583","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-10-27 13:51:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"14d3655582000e6422d9eb1046cd3075d768dfda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 13:52:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7527f830707cec4d821193abbd8ecf4637b8d685","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 15:10:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ecec3154084774960279cf7842ed74 : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7310dc61fe914c82ac284b64d59e825c : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0171776a3db4d2ca4d4322081fce952 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63d04e3ad77e44d1b588f5d94d5ccf79 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b75f1f642fdc4b1d97606aab0c7c7735 : SUCCESS in 5m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50904a8a654d41e49ce6411d9c641e03 : SUCCESS in 1h 16m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9bec47d5ffa8404fa6d0181f4a6f557b : SUCCESS in 38m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ba36dda49fe4dc0b48de17aa320d89d : SUCCESS in 52m 06s","accounts_in_message":[],"_revision_number":2},{"id":"9d2feb06562a85f6efa442112203517cc04a2355","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 15:10:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fb7ba34d925b7d2920a9fa1754dbd8d9103d5c38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 15:12:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85ddb35ba6c44d81afa63b68448e4429 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5195d8c965f8b7536a4ea32ef7c64a30b6c4e28","revisions":{"8c3fbf33c2729f46754dafc8575810fb399efaba":{"kind":"REWORK","_number":1,"created":"2021-10-12 05:10:05.000000000","uploader":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"ref":"refs/changes/57/813557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/813557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/813557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/813557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/813557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/813557/1"}}},"commit":{"parents":[{"commit":"e686f3d0d36a519a2efe80e4273749e08c80e521","subject":"Add default values in port forwarding port ranges definitions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e686f3d0d36a519a2efe80e4273749e08c80e521"}]}],"author":{"name":"Manu B","email":"manu.b@est.tech","date":"2021-10-12 05:09:01.000000000","tz":0},"committer":{"name":"Manu B","email":"manu.b@est.tech","date":"2021-10-12 05:09:01.000000000","tz":0},"subject":"Addition of status and name field to bgp_associations API","message":"Addition of status and name field to bgp_associations API\n\nSigned-off-by: Manu B \u003cmanu.b@est.tech\u003e\nChange-Id: I3478950acb99f686a5ae3758729862b48cc9bd7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8c3fbf33c2729f46754dafc8575810fb399efaba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8c3fbf33c2729f46754dafc8575810fb399efaba"}]},"branch":"refs/heads/master"},"d5195d8c965f8b7536a4ea32ef7c64a30b6c4e28":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-25 09:13:35.000000000","uploader":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"ref":"refs/changes/57/813557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/813557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/813557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/813557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/813557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/813557/2"}}},"commit":{"parents":[{"commit":"e686f3d0d36a519a2efe80e4273749e08c80e521","subject":"Add default values in port forwarding port ranges definitions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e686f3d0d36a519a2efe80e4273749e08c80e521"}]}],"author":{"name":"Manu B","email":"manu.b@est.tech","date":"2021-10-12 05:09:01.000000000","tz":0},"committer":{"name":"Manu B","email":"manu.b@est.tech","date":"2021-10-25 09:13:18.000000000","tz":0},"subject":"Addition of status and name field to bgp_associations API","message":"Addition of status and name field to bgp_associations API\n\nRelated-Bug: #1921461\n\nSigned-off-by: Manu B \u003cmanu.b@est.tech\u003e\nChange-Id: I3478950acb99f686a5ae3758729862b48cc9bd7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d5195d8c965f8b7536a4ea32ef7c64a30b6c4e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d5195d8c965f8b7536a4ea32ef7c64a30b6c4e28"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
