)]}'
{"id":"openstack%2Fneutron-lib~816447","triplet_id":"openstack%2Fneutron-lib~master~Ie6fffb7829ac1f212f4c4e40fc2e5161f2ccd6fe","project":"openstack/neutron-lib","branch":"master","topic":"bp/qos-minimum-guaranteed-packet-rate","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-04 12:27:45.000000000","reason":"Change was submitted"},"33652":{"account":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"last_update":"2021-11-04 08:47:11.000000000","reason":"Przemyslaw Szczerbik replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2021-11-04 12:27:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie6fffb7829ac1f212f4c4e40fc2e5161f2ccd6fe","subject":"Move RULE_TYPE_MINIMUM_PACKET_RATE to neutron-lib","status":"MERGED","created":"2021-11-03 10:24:17.000000000","updated":"2021-11-04 12:29:43.000000000","submitted":"2021-11-04 12:27:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816447-bp/qos-minimum-guaranteed-packet-rate","meta_rev_id":"36e35a3e45b8038e218e6f4f7d2e2beea268b7ec","_number":816447,"virtual_id_number":816447,"owner":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-04 12:27:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-11-03 14:54:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-11-03 11:04:39.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-11-03 14:54:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-03 10:25:16.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-11-03 10:25:16.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-11-03 11:04:39.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-11-03 11:54:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-03 14:54:12.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"d37d92919c55a7954407dd512882271a9ffc17e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-03 10:24:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46b5b72cf6b67633dbab2b44e15a201f64a94518","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-03 11:04:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9283f6a5e1b1cdb68ec377d20f20b0bc5e232646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 11:54:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b66e7ab4a944b7b5346a0a7f5ce51b : SUCCESS in 7m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50a65dee33864f85b5f369795910a3b3 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e25682e4ded47af8147f5f453b14805 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/499942fd25484eb8a563888fadab95fb : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff47c8c3928642e08b73adc2579330a2 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a4a7af589d54bd2a01a3f1ff6624082 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce50ddf7351c4465949792812f564dfa : SUCCESS in 4m 45s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ccbde55deed4bd3869525a6dd5b22c3 : SUCCESS in 1h 24m 33s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/67159df4c1e346019d051c250f58182a : SUCCESS in 45m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7ab0a5db14a04d1cb6cf288dfdb64dab : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e0be6169d069eb96bbd6d1d232abfcac3684d87f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-03 14:54:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"024356413c6fc69b72413cefa1fe6ef281c21144","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 14:55:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d5d1daee6ed46ad64f0ce583eadf1c2a2db3062","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 16:27:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba28ae12d3d4aa9a11c402ae469262d : SUCCESS in 6m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce526f8926c1406684edd902898bae26 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/356e588f9d89409fb302f366bb66b6b2 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1c8c7a0b3694a79aadc15df950d30dd : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b77823f08a840038f54436df185e11f : SUCCESS in 4m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fac10710a67404b9ea67de71e16c5b1 : SUCCESS in 1h 26m 56s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d5e80e4178bb45b4ac45b50e97eeee5c : FAILURE in 44m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/024ffd8876984067be1775c475e2ba53 : SUCCESS in 49m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ac48cc74fc7a10595ba42470821c7638502ec1a5","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-04 06:27:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0511b8c781841c885452d9e10f1b74f9bdc0bae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 08:05:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa360e9e8d34444683bb011df5fa96ec : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61bf59c8444846cfaddff345f304ed11 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5bb1c39e24348ef9f57b6534e7dd095 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cdeb360506e406eb2a7dee64cf62eba : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0f8c83b84a945b9ae9592273e299441 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaed7456b5e44dd0a4cd4e3bb80d8f09 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a1801821fa74a33a07a2ec6416e3c44 : SUCCESS in 5m 23s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d913f24e9b8a43ff99337a19df3418c5 : FAILURE in 1h 35m 52s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/590d7387ef504aa39f6ddf6ca9e4333b : SUCCESS in 40m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/48986c438e5d40fbbd281b3dbe431e5b : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d9d9b7024c09afa7c6fce3a74e8a69d431112d38","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-04 08:47:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"169124f4a343ba83fb58c6578268bc137a7dd9f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 10:23:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709f26bf95c8486aa1599ab8528deeac : SUCCESS in 6m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c86e10321d1443e3a1ffb0b8da9ca799 : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56564dadac8d463a86bb430ea4222a7e : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/683a2e84eb2343cd856272f7d324ba4e : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/527cd78e67744828a8b9510084973ebe : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5767efa361714920859911522a046e2a : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/087f9a0b513e46618be9daa8e8b8f470 : SUCCESS in 4m 52s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af1221e5b4bc4dd5852c7287d5f14395 : SUCCESS in 1h 31m 40s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3334c74f30f3459189729f03c10a6958 : SUCCESS in 45m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c8d45e7b1aff448a9928ca07ffde804a : SUCCESS in 49m 20s","accounts_in_message":[],"_revision_number":1},{"id":"9d81bb7a18795fd6aea33d8a7cb4fedeee1c4b32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 10:24:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff0853eb12888c9b7a723f0c6c7dee595c6d4faa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 12:27:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cfef257891c457da4a0aee25c0c6533 : SUCCESS in 6m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa200b175fbd4b8d9ca02334eeb1e236 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0622a46a7aee4304b4ef4c92fe6bf70d : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6559657062354337bc751206180fee50 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6442a8c268294d27be0a8119720ad0e2 : SUCCESS in 4m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5ca52c416bb4795b03c585cb012d966 : SUCCESS in 1h 57m 31s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6d87246355f84af2a50ac5dbceec4fa1 : SUCCESS in 46m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e68909953014ef7a625d803c5aa8cb8 : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3de867b6e465a81e75868ddbc3f132495c5d2163","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 12:27:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"36e35a3e45b8038e218e6f4f7d2e2beea268b7ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 12:29:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f8b297e6f948a2ac27a87ed2a36f04 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b2fe94b2ccb49a7906fbda8cfe48892cf37833c","revisions":{"6b2fe94b2ccb49a7906fbda8cfe48892cf37833c":{"kind":"REWORK","_number":1,"created":"2021-11-03 10:24:17.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/47/816447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/47/816447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/47/816447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/47/816447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/47/816447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/47/816447/1"}}},"commit":{"parents":[{"commit":"4eb64ecf3227ec57662acbcb045926ae7674438e","subject":"Merge \"Move QosPlacementAllocationUpdateConflict to neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4eb64ecf3227ec57662acbcb045926ae7674438e"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-11-03 10:18:00.000000000","tz":60},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-11-03 10:18:27.000000000","tz":60},"subject":"Move RULE_TYPE_MINIMUM_PACKET_RATE to neutron-lib","message":"Move RULE_TYPE_MINIMUM_PACKET_RATE to neutron-lib\n\nAddresses TODO comment from:\n\nneutron/services/qos/constants.py\n\nRelated-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie6fffb7829ac1f212f4c4e40fc2e5161f2ccd6fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6b2fe94b2ccb49a7906fbda8cfe48892cf37833c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6b2fe94b2ccb49a7906fbda8cfe48892cf37833c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
