)]}'
{"id":"openstack%2Fneutron-lib~816468","triplet_id":"openstack%2Fneutron-lib~master~I1ec3185becf03bdc208b0af8c00d2794113a88d5","project":"openstack/neutron-lib","branch":"master","topic":"bug/1930200","attention_set":{},"removed_from_attention_set":{"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2021-11-11 09:00:50.000000000","reason":"Oleg Bondarev replied on the change"}},"hashtags":[],"change_id":"I1ec3185becf03bdc208b0af8c00d2794113a88d5","subject":"Add Local IP constants needed for policy checks","status":"MERGED","created":"2021-11-03 11:54:07.000000000","updated":"2021-11-11 12:26:45.000000000","submitted":"2021-11-11 12:25:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816468-bug/1930200","meta_rev_id":"1f4b48b6f8807df291e71a00a86d51e49d716507","_number":816468,"virtual_id_number":816468,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-11 12:25:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-11-03 14:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-11-09 17:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-10 08:45:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-11-09 17:02:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-03 13:13:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-03 14:58:24.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-11-09 17:02:07.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-11-10 08:45:46.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"a3cfe4fff2c38a9f6253578057b424fdcd8729e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-03 11:54:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2321b8fe011e435f8892d9061388c9a35c1ed3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 13:13:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f201bf0324d455fbb993a679aca5227 : SUCCESS in 8m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf092acb0221437a844dc3d1f66ee878 : SUCCESS in 6m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f556bd290b3d428c938c4bc3684e774a : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4d2a4c661e843098382d21fa4835137 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/823fb309df07479788ac62ec04b42f90 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85603caa9cb54b4ba2ffedda2fba0b16 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a1e8b043ea94931bf69e88349a14a03 : SUCCESS in 6m 02s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb653232e3424fcea5bd01541988995a : SUCCESS in 1h 16m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3d9d71f213fd48f7a4f68ed066211278 : SUCCESS in 49m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9cfbe51f895e453daa55af6b86c79f33 : SUCCESS in 54m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4cb8fecc9f7198d37985ab1332eddf0f7c61081f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-03 14:58:24.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ee48f51420513c999a13c7f4bee4047162171ca","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-09 17:02:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0c4e8f43ec8238175960a87e64a6be40f3711fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 17:05:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16334dc6a01404ecc0a62e8bc7723a2b22a92f04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 18:33:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8041001296474a94bf4351ead66024a0 : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1f8da6106464c3ebc4f8bb5cf340bb4 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c183f7d68147d9a5b43a359972c9bd : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49c69c72b37042b0902acb5361efd95f : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8009e1a3cd549a99d284584395f6c1f : SUCCESS in 5m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab3087ad39d24ddab7bcc68a4b616b9b : SUCCESS in 1h 27m 11s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/57b2b8507e82426abbadce526cba7b02 : SUCCESS in 40m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c9ad3c7d66b14dc88f3ab4b8adfef344 : FAILURE in 48m 15s","accounts_in_message":[],"_revision_number":1},{"id":"587a3b8d09a8c64439b627f3941325ce9f97b2ef","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-09 18:53:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"731f6224832ec7b1770b992108cb157af0887197","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 20:11:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c2b4bb535374f1da1d8595e6e252b5e : SUCCESS in 10m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cc4572435604943afd96edce5578793 : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2935a33bfdcf42bfa16097ccfdb6fca1 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/028aa71e812b4cfd89951807e5f1e79a : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06d2f9d259494ac39fd1856bf68e0001 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fbde01e7d00475bab62e7f1ef680895 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28ae00834bd542d9b9117ae27acd1a0a : SUCCESS in 5m 30s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19c5811ec04a4e259236334223b20bcb : SUCCESS in 1h 16m 33s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8fcd8508c2a049a19f1343ef496266c6 : SUCCESS in 38m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba38835ab54342279cdde25747654446 : SUCCESS in 46m 19s","accounts_in_message":[],"_revision_number":1},{"id":"70691151271e66441a567e52b7eda5702822c1e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 20:12:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78cd889b984950ef8fb49b77678f3c1c49f81038","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 21:29:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff3a92e0c25540289becf021a422cc4d : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3597911778b84a71ba775cdf7cdf0f0d : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b933387b9e4065a0795c880f3d1264 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2921853d74b04d61bbb2e36204d5c295 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91f36c7ce45e44b8a086dc511bcd1b31 : SUCCESS in 4m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9dbb80ee804a42e8902226a0591011f0 : SUCCESS in 1h 15m 10s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/375a1ad90f8443d2980e23daac41a72f : SUCCESS in 38m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3430b76bfe8543dda359667cea9eb761 : FAILURE in 46m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e8881c49ccca36e81c563194058c56b2d4ffc380","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-10 06:57:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40c3267371f02cfb7337bb6b6ac55cac9552d75f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 08:21:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a609f289da804960a0b689459f1d07e3 : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d87bc6b27c004ed2a7a99dd610f8a785 : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23757dbf81144bd892c97f5968336d58 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15a2a86bf9c64a87b69fcbfee0b48125 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b01ea656f2b445db1c45c9c01a5ac11 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40c5f20bf8ef44848ab873b28431f22c : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/318c6924b74749bba1a92293fb3b45bc : SUCCESS in 4m 31s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/303384b161044aaf80256d3e3eb1fb42 : SUCCESS in 1h 21m 43s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/540c3673ae85444ab2b74bc18509bda7 : SUCCESS in 41m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7efb583e06a04ce09a5446ba280195fc : SUCCESS in 46m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c3a830a8499a0a6edd8d5dff825bd9fa2dbcafae","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-11-10 08:45:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f45c3cee02f8e06ab90730935ba2ecabc209ddcd","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-10 11:13:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4bd666a76bd479a361c09a2c392ed6f2be822cbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 12:52:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba8e269fc214dde90b3e19c29cf83aa : SUCCESS in 5m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c496acd729f4c78859579edd168037d : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10551679ecda4e578397d25c86364d92 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fb0ee23d5004cccaf48dc484da2bb6e : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2a6b92f473b4e9c98794d53eb553704 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ad2a8f6149140f6b67139c3c3c8fc13 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a812a307bede413c932dfa4808b1734f : SUCCESS in 4m 32s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa9dbc7a7f4f4e03866cf4cde593fd0b : SUCCESS in 1h 35m 31s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d0a6b9b4df7240abac08621e8918c6c1 : SUCCESS in 40m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/31682461c4d94542b8dd0918c293a28f : FAILURE in 48m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9374ba66bfba50e02876ea0b07047e04d50deb07","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-10 13:11:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2bac102123e3663be5cf9dcfd2ba2f88b8c2890d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 14:34:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40fef7d68d4d407a88ac36cfe2053a41 : SUCCESS in 8m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7383bb5f9fa41e9b253337e9a9d279e : SUCCESS in 6m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72d2b35e31c34a0eaaf6ffd6de273014 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bc4f5d8533b46da9a72b8956c8f8374 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d27a2a203ed4bcd98e77fd2b3466548 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/204c8557c65f43989083974f327fab45 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76c4ba3e71924eb88a509955708499a5 : SUCCESS in 5m 46s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e1be0209f7446358e0457476dca5450 : SUCCESS in 1h 20m 17s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ad66996f30064549a5fc8952a95600a6 : SUCCESS in 46m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef92451e42784d8b9daf3c51e94d32a0 : FAILURE in 47m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d1631ffeb0b5570f44e159e736abf44c391005c2","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-10 14:37:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30a058d7b804913e01b2f48cd38251847105f517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 17:19:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e70c336b4c49299218a70caa64f217 : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a948894c29c4a08897e87e6f9e94622 : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e65240bbd4e48a5914112eef83408f5 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efa086cd63c74710a755fa8711926914 : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04f96aa55c554e0fa871ab1566f13edd : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0eb28ff9b344b5dbb7358ff9b949384 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e51c7cd97c1e437780a1b2e36fe5cf96 : SUCCESS in 6m 15s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/235afffdc91649d39867124e5e5b4d64 : SUCCESS in 1h 26m 54s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/33b81a24eeb94c4db287207b48ca5f1a : SUCCESS in 51m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7abf6bea0f0349dfa78b3906699fceec : FAILURE in 55m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b184626846aa29308eb15ed296e3c43545a7f3d7","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-11 07:38:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be15af4f73e175a8f21c178565dbdd3443f6a4e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 08:37:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64c1bfa791d74ca98a9d600006c5f0a8 : SUCCESS in 5m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fced271ee0e941ac89b748e101868475 : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45f3dbaaf5f44ccd838ccdfd6578f580 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fbbf2a58c984683964cba02518a7e37 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/020552821eb140f1b8955efcf467e237 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c7b21556b9e4e03ad11207d512e6e50 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62228864a6434fe8b068ca07e2827c05 : SUCCESS in 4m 31s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01cd632e0df743ce8d2c71bc9b538cc0 : FAILURE in 17m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ffc8188f708242d49e675d6c2bd9795a : FAILURE in 20m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/213dbfc13f13418abfd9a8d91bf07990 : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5bea96d3f65442cbfa2876f0c84b573e65edb179","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-11 09:00:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"916b3132563f92e4f25137b8a5afdc0b02883259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 10:44:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9915ceadfb634d91b6202500cb964e56 : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59129100d544404d9083133e9ca04c1b : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c85c7ad5abca44f9b7e0efbd6c606ee1 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b033d0ff68af4325943217243997eaaa : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64c47e99cd21489bb6f22af40d424c4e : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6837bb2fb6b4a0fbf2856b408888327 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26e6754599c746ac9e69f3e36e36682f : SUCCESS in 5m 07s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf50552dd0e4407eb71841e7b0607ce3 : SUCCESS in 1h 38m 32s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/506cb04a65a64e00900b707ccf55f383 : SUCCESS in 44m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/db7a641fa1054f75985380d8f095ced2 : SUCCESS in 50m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3564b7982ab1a1004091ae52ec02e4ed54059b17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 10:45:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a0cd564e3a3e3451ae85fcd2e36177be9ba90bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 12:25:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ab837e25a0247d9afc80dcc329adff9 : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c222e7a9c4594b92869f220043087c99 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd6f0e48cb4f438f98de56553a7c7da3 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05a1ef96157f45e1860c51c09361ecb5 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b7e308ad3ec4d159ecb40ef4a625f82 : SUCCESS in 4m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8dfe65428e5a43a3ad82bc1e00f59eb3 : SUCCESS in 1h 38m 22s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/23b97aa2f7084f4c8b8e555b710ffc2d : SUCCESS in 34m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a16fe58a846540028dec689398555fe1 : SUCCESS in 43m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9ca6cf8e9d3193639f6560171a86c2a8a7a5af09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 12:25:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1f4b48b6f8807df291e71a00a86d51e49d716507","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 12:26:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3607a6736949ab95cef87a5de7281e : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d4da16374ff378543f03d85dbf5d564c8f38391","revisions":{"1d4da16374ff378543f03d85dbf5d564c8f38391":{"kind":"REWORK","_number":1,"created":"2021-11-03 11:54:07.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/68/816468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/68/816468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/816468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/816468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/816468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/68/816468/1"}}},"commit":{"parents":[{"commit":"4eb64ecf3227ec57662acbcb045926ae7674438e","subject":"Merge \"Move QosPlacementAllocationUpdateConflict to neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4eb64ecf3227ec57662acbcb045926ae7674438e"}]}],"author":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-09-29 09:22:38.000000000","tz":480},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-11-03 11:53:06.000000000","tz":180},"subject":"Add Local IP constants needed for policy checks","message":"Add Local IP constants needed for policy checks\n\nAdditionally adds required small fixes to API def\n\nPartial-Bug: #1930200\nChange-Id: I1ec3185becf03bdc208b0af8c00d2794113a88d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1d4da16374ff378543f03d85dbf5d564c8f38391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1d4da16374ff378543f03d85dbf5d564c8f38391"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
