)]}'
{"id":"openstack%2Fneutron-lib~817528","triplet_id":"openstack%2Fneutron-lib~stable%2Fxena~I9a0df4d461c0e73ba07ca18648f16a982a2e267b","project":"openstack/neutron-lib","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-24 16:39:52.000000000","reason":"Rodolfo Alonso replied on the change"},"33652":{"account":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"last_update":"2021-11-25 22:35:48.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2021-11-25 22:35:48.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-11-25 21:22:49.000000000","reason":"Slawek Kaplonski replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2021-11-25 08:46:30.000000000","reason":"Lajos Katona replied on the change"}},"hashtags":[],"change_id":"I9a0df4d461c0e73ba07ca18648f16a982a2e267b","subject":"[stable-only] Fix allocation update for min bw rule with different direction","status":"MERGED","created":"2021-11-11 10:13:47.000000000","updated":"2021-11-25 22:37:09.000000000","submitted":"2021-11-25 22:35:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817528","meta_rev_id":"c451142cd17c52ed935d38b1d7c14156e38b7a2b","_number":817528,"virtual_id_number":817528,"owner":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-25 22:35:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-11-24 16:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-11-25 21:22:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-11-25 08:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-11-25 21:22:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-11 10:18:53.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-11-11 10:18:53.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-11-11 10:18:53.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-11-11 10:18:53.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-11-11 11:38:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27ac3fc5b0036eef0d6e81a94d6e8bc2496a0deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-11 10:13:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc8f0c588b093aa6457ca9320a2882a7edc3a531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 11:38:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c8f1aa4de8e4ff9820c479d538468f1 : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bea6d3638bd4e29a9b8f0fc6bc19276 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09d01611642f4a34ac3abf4d11217f33 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffd02671da9b4112aaa74a7f80a26644 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8848df77b87246f5bbc5f9829aa1210f : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78b9ab56bae244f39a0cac035d38d20d : SUCCESS in 6m 43s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1762dbf5e61c4c52a14ec69a06b8dd8e : SUCCESS in 1h 20m 32s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4aa4e7633e484d48ab396ac949bd6deb : SUCCESS in 1h 05m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5f8fd6dbb71c4351a952def90b60b891 : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":1},{"id":"712d76d346711a57ed442c82f58f5b985fd83512","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-15 15:05:14.000000000","message":"Topic bp/qos-minimum-guaranteed-packet-rate removed","accounts_in_message":[],"_revision_number":1},{"id":"61742440d7451ffa631bf3ffbefcfe3fcce388b3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-19 12:00:27.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e5e0e79c68d6385f8d340b759077fc1a4760903","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-19 14:33:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20e0e4956e2db897564310051622ee1a824e8980","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-19 14:40:57.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4cee46b7b0ffc4e724f940278441ced2f6cc9278","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-19 14:43:54.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"520cfaf1ecd68a9dc1f04a4ffcded02bc8c048cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 16:15:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aa917c7c1774b7796f1be7f732d295a : SUCCESS in 5m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14380f3be6d340ca9c95704c28f047e4 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c875db0f700843d79a6ff22ec488725b : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8440c0460ec44861819de0b6927199d2 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d37c6e2380e3480c89c4f86fe50a0465 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95374c29ec124268ae345fb07afbb703 : SUCCESS in 4m 17s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c58ea0fb919c489280bff3c82050e255 : SUCCESS in 1h 28m 11s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c663c44c185d495ab83e4669560ae69f : SUCCESS in 54m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/325bbd5d4f114312a412e7c870148ada : SUCCESS in 55m 12s","accounts_in_message":[],"_revision_number":2},{"id":"24cb4799bcce64c46e79c69fbc74f64c88252b09","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-24 16:39:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ca5be8d18d903edf10baac919cfbb577165a879b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-25 08:46:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e2d62a309127998d5880c61d0a63cded85bd042e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-25 21:22:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1c46feedcf0911eda17e6662d76212c81137debe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 21:23:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b927714ec6404b882777c19cf8fa8c8a265203c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 22:35:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9b915eb8ff4153a76e1fda5ad54ada : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43a2201a4a0444d6b4b2c525504a2077 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/219970cda6e7436ab14e71c070a801c3 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e3644b6b3784819963f5bb0c0c167be : SUCCESS in 4m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17e6170e59c34998900c1e9ddb6fde3f : SUCCESS in 1h 11m 00s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/de9b0cb38acd4ee787cef8659334d8e3 : SUCCESS in 41m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/22d2bc315307498283951a1cf99f01bc : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":2},{"id":"8b6573942f1ebcf7e77e980078cd7eab870c5f92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 22:35:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c451142cd17c52ed935d38b1d7c14156e38b7a2b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 22:37:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ef6f2d726b4010ba6a6d9c09256520 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"997be4d13b183b5f9bac1462da5aeaec0c44714d","revisions":{"d9c89cd72b5c546b12729aac2f72a4f0fa1012c4":{"kind":"REWORK","_number":1,"created":"2021-11-11 10:13:47.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/28/817528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/28/817528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/817528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/817528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/817528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/28/817528/1"}}},"commit":{"parents":[{"commit":"19cf80f06c43b56d31c1d6b0e3cc1d1b4565ad96","subject":"Dropping lower constraints testing (stable Xena)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/19cf80f06c43b56d31c1d6b0e3cc1d1b4565ad96"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-11-02 10:24:44.000000000","tz":60},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-11-11 10:03:49.000000000","tz":60},"subject":"Fix allocation update for min bw rule with different direction","message":"Fix allocation update for min bw rule with different direction\n\nNeutron API supports QoS policy change on a bound port. The new QoS\npolicy can contain a different set of rules, or different values. In\ncase of QoS minimum bandwidth rule, it can be min_kbps or direction\nattribute. If the new QoS minimum bandwidth rule is different than the\none in previous QoS policy, Placement allocation has to be updated.\n\nAn issue arises when QoS minimum bandwidth rule direction is changed.\nIngress and egress directions are represented with separate resource\nclasses: NET_BW_IGR_KILOBIT_PER_SEC and NET_BW_EGR_KILOBIT_PER_SEC\nrespectively. During direction change, resource class representing old\ndirection is freed and a resource class representing the new direction\nis requested. update_qos_minbw_allocation() implementation always\nexpects requested resource class to be present in the allocation that\nis being updated, but that assumption is wrong. If current Placement\nallocation record doesn\u0027t contain particular resource class, we need to\nuse 0 as a default value.\n\nSimilarly, we should defer deletion of RP from request body until all\nitems from minbw_alloc_diff are processed. Only then we can be certain\nthat this RR is not going to have any resources.\n\nThis commit is based on 47a08e36bf864806ca71ea77ffcf6602d4e5db04.\n\nPartial-Bug: #1943724\nChange-Id: I9a0df4d461c0e73ba07ca18648f16a982a2e267b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d9c89cd72b5c546b12729aac2f72a4f0fa1012c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d9c89cd72b5c546b12729aac2f72a4f0fa1012c4"}]},"branch":"refs/heads/stable/xena"},"997be4d13b183b5f9bac1462da5aeaec0c44714d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-19 14:40:57.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/28/817528/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/28/817528/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/817528/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/817528/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/817528/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/28/817528/2"}}},"commit":{"parents":[{"commit":"19cf80f06c43b56d31c1d6b0e3cc1d1b4565ad96","subject":"Dropping lower constraints testing (stable Xena)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/19cf80f06c43b56d31c1d6b0e3cc1d1b4565ad96"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-11-02 10:24:44.000000000","tz":60},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-11-19 14:40:57.000000000","tz":0},"subject":"[stable-only] Fix allocation update for min bw rule with different direction","message":"[stable-only] Fix allocation update for min bw rule with different direction\n\nNeutron API supports QoS policy change on a bound port. The new QoS\npolicy can contain a different set of rules, or different values. In\ncase of QoS minimum bandwidth rule, it can be min_kbps or direction\nattribute. If the new QoS minimum bandwidth rule is different than the\none in previous QoS policy, Placement allocation has to be updated.\n\nAn issue arises when QoS minimum bandwidth rule direction is changed.\nIngress and egress directions are represented with separate resource\nclasses: NET_BW_IGR_KILOBIT_PER_SEC and NET_BW_EGR_KILOBIT_PER_SEC\nrespectively. During direction change, resource class representing old\ndirection is freed and a resource class representing the new direction\nis requested. update_qos_minbw_allocation() implementation always\nexpects requested resource class to be present in the allocation that\nis being updated, but that assumption is wrong. If current Placement\nallocation record doesn\u0027t contain particular resource class, we need to\nuse 0 as a default value.\n\nSimilarly, we should defer deletion of RP from request body until all\nitems from minbw_alloc_diff are processed. Only then we can be certain\nthat this RR is not going to have any resources.\n\nThis commit is based on 47a08e36bf864806ca71ea77ffcf6602d4e5db04.\n\nPartial-Bug: #1943724\nChange-Id: I9a0df4d461c0e73ba07ca18648f16a982a2e267b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/997be4d13b183b5f9bac1462da5aeaec0c44714d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/997be4d13b183b5f9bac1462da5aeaec0c44714d"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
