)]}'
{"id":"openstack%2Fneutron-lib~818536","triplet_id":"openstack%2Fneutron-lib~master~I7184c058ea59a1dceefda7ce4488df4d64c71111","project":"openstack/neutron-lib","branch":"master","topic":"bug/1950454","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-12-13 10:44:05.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2021-11-29 11:07:36.000000000","reason":"Lajos Katona replied on the change"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2021-12-13 10:44:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7184c058ea59a1dceefda7ce4488df4d64c71111","subject":"Add new method \"update_network\" to \"L3AgentExtension\".","status":"MERGED","created":"2021-11-19 11:17:09.000000000","updated":"2021-12-13 10:45:40.000000000","submitted":"2021-12-13 10:44:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818536-bug/1950454","meta_rev_id":"2626beb5d82f9f37ea899cf99c99800602e60664","_number":818536,"virtual_id_number":818536,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-13 10:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-12-13 09:18:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-12-13 10:45:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-12-13 09:18:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 12:18:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-19 14:26:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2021-11-23 08:33:29.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-11-29 10:30:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-12-13 09:18:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"4d5b77b9e77bb5f1ead33744adba1c7722f6e716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-19 11:17:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5e2721506ba5f4bad34dd4ea8dbedb8ed373057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 12:18:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e27c8ea497f472b8dad14e24d7a97aa : SUCCESS in 7m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecb4262810d340fd9afacf6c36ee3c12 : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e2774fdc2284b54897279dae8875cb2 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08e1cee4fc6d4230940284dfde1e5ae2 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3166e1879d34450a0f4ea9380fb73f4 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf4c62e7a9ed49b0a8b752d419fdc7eb : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/341eaa7f648b4e1fac95d7999746e04d : SUCCESS in 4m 28s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ad92d42c5ff44c18a799d6ca00213d2 : FAILURE in 15m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad5d952a7ba449c6903b72c0f67ee878 : SUCCESS in 3m 34s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e036859f98cc469aaee46363fe2ebb6f : SUCCESS in 41m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f33f079fbc804d76a97680e4eb2695bf : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":1},{"id":"90391b2487c7e095eb0072edb20b07ecb89b26ef","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-11-19 14:26:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0ce5cd2ac99db56e23d84d9bd21eb6fda746417","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-23 08:33:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"097a742359cbf80b731f65f72ca891f92e19a8b8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-23 08:34:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6acc76d49fe65746589e074ae647d0d39cc6f256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 09:54:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8d5b6d8f22e4f35b5f24ad6e38bcc4c : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/155ef45f2f964946b46b1bd88267719e : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ead9c264ed7435592a7c7c12be9eb0d : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be65b9d52db64023a7de322c8073a481 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0327af5e56e8412abc8447becb2bc6b3 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68f87588f1cf4e9d82bb98222aced17c : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a2a5fd14ca74ca1be3859fd098737dc : SUCCESS in 4m 13s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd83539b5e8b4e358c59e471d84eabac : SUCCESS in 1h 17m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab2daaad99884e1db19c78b9f246f167 : SUCCESS in 3m 45s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1999d51132c54dae960f331f0f4abf18 : SUCCESS in 38m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c0cff87dbbc46b4b2f91d73ef4de136 : SUCCESS in 44m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b8c979afbf4d4d4540d3552a5b564d5f61f9fbec","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-24 16:58:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b13fb665d5880eab954e79978563a2e35997acac","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-29 10:30:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ae30520a1245d303a9ac54bc5504ca3d9e944c1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-29 10:43:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94d18473df61aa6323c21c9844532c6ea06f94ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-29 10:46:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"358e413ce70eebe06cd6622b6d1cc897d3cbc452","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-29 11:07:36.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c3ae0ba1a7c2b67ebb93ddb942565bec90314d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 12:07:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2d79024ab4b46429b775edf8cd6ee95 : SUCCESS in 6m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7256b10066c54e7897c92f2510bf6325 : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/405e420e287242679d777da5829c8e93 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/569f6866cdd9408ca85e88bd975dfc39 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2a58c3e47214f95a9e874f5c9a6bca5 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97027ee634e14f848a97e2dc8f4c3ced : SUCCESS in 5m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55beb423879f4a18a9140b35a855f714 : SUCCESS in 1h 18m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbb1860f37884916b147aff260c81bae : SUCCESS in 4m 22s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/db8ff67d853143db9178c717a97226b9 : SUCCESS in 59m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cfbe60e856264f20b7f54835738a9644 : FAILURE in 33m 32s","accounts_in_message":[],"_revision_number":2},{"id":"ea2867697a5dd3909b979ceff1fe9161d66060d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-10 10:43:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0b4ba4143f66672cf74c0434040727dba9a633ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 12:01:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb3c18fa7fd444a9dc756650bf5d59a : SUCCESS in 6m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78c142843fa74570ba58dd78b2417557 : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aebdfa3079de45aab3f3e5abe38e696c : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c58a860c1744c0fbf43bcc8782299ea : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46a9aebf56ca49808a720eac2cb72083 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14d16b3514594aa2aef7e3b733181044 : SUCCESS in 5m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56aa7b58c126475cb06a1cf05c7f0817 : SUCCESS in 1h 16m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a211eeb51a91492dae29d772a8a1801e : SUCCESS in 4m 01s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/94a623927e5b4a878a485c006ffd5aed : SUCCESS in 47m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f8043e6777d841d9869b3e46b3f7cb93 : SUCCESS in 42m 18s","accounts_in_message":[],"_revision_number":3},{"id":"054961b8b7c8d2eaff86c310061ef264e57162b4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-13 09:18:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1ec8e17228370d270591cd5a1b297b432b4cd2c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 09:18:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a683587bfc7f44d3aa42323a65260dbd05e4899f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 10:44:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63fa4b317de14e459b5b94ce3510f6a6 : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56cb1c1711cb4d7b9974a30dadf936e6 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15f2b9ef45c049d1a8c1c6bcbe3e1ef2 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c1eda8eb29b4e00b74ea848b5540075 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a60a4eba502a434abf2699a7dd39337f : SUCCESS in 4m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/795a4da6939a4233909fcf04814a3fb7 : SUCCESS in 1h 23m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15caf98737f4483fa150e75de02880e5 : SUCCESS in 3m 54s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c2fb69a2fe5c4f73be46706a29e49b2f : SUCCESS in 35m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/96b76b97d1304630a06b726af4123340 : SUCCESS in 44m 54s","accounts_in_message":[],"_revision_number":3},{"id":"29f6e4a342db0659ca6c76308140dc0f61399312","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 10:44:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"982e7589cacb38f0087026ead737f494512490c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 10:45:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e88df4a583c441f0b21b0073958ac53c : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f77686c256134276a12950180d42a016 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4807bacab4ad6320cbd6b3c2c164bad78f5f42d5","revisions":{"4187238484424d37064ccbcf433dc5696c48813c":{"kind":"REWORK","_number":1,"created":"2021-11-19 11:17:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/36/818536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/36/818536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/818536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/818536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/818536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/36/818536/1"}}},"commit":{"parents":[{"commit":"b35ba2fe95583e156192333b03a0b4dbeab8ef39","subject":"Merge \"Add Local IP constants needed for policy checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b35ba2fe95583e156192333b03a0b4dbeab8ef39"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-19 11:14:53.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-19 11:14:53.000000000","tz":0},"subject":"Add new method \"update_network\" to \"L3AgentExtension\".","message":"Add new method \"update_network\" to \"L3AgentExtension\".\n\nThis new method will extend the L3 agent extensions functionality,\nmaking them aware of any network change.\n\nChange-Id: I7184c058ea59a1dceefda7ce4488df4d64c71111\nRelated-Bug: #1950454\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4187238484424d37064ccbcf433dc5696c48813c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4187238484424d37064ccbcf433dc5696c48813c"}]},"branch":"refs/heads/master"},"1cd3060a10bd2dc2ac9b38f0f29f96062cf933ee":{"kind":"REWORK","_number":2,"created":"2021-11-29 10:46:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/36/818536/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/36/818536/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/818536/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/818536/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/818536/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/36/818536/2"}}},"commit":{"parents":[{"commit":"b35ba2fe95583e156192333b03a0b4dbeab8ef39","subject":"Merge \"Add Local IP constants needed for policy checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b35ba2fe95583e156192333b03a0b4dbeab8ef39"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-19 11:14:53.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-29 10:45:49.000000000","tz":0},"subject":"Add new method \"update_network\" to \"L3AgentExtension\".","message":"Add new method \"update_network\" to \"L3AgentExtension\".\n\nThis new method will extend the L3 agent extensions functionality,\nmaking them aware of any network change.\n\nChange-Id: I7184c058ea59a1dceefda7ce4488df4d64c71111\nRelated-Bug: #1950454\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1cd3060a10bd2dc2ac9b38f0f29f96062cf933ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1cd3060a10bd2dc2ac9b38f0f29f96062cf933ee"}]},"branch":"refs/heads/master"},"4807bacab4ad6320cbd6b3c2c164bad78f5f42d5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-12-10 10:43:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/36/818536/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/36/818536/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/818536/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/818536/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/818536/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/36/818536/3"}}},"commit":{"parents":[{"commit":"fab7a88062c0aec90e02b4fe7666af1bf79716db","subject":"Merge \"Exception for BFD: BfdMonitorInUse\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fab7a88062c0aec90e02b4fe7666af1bf79716db"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-19 11:14:53.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-12-10 10:43:03.000000000","tz":0},"subject":"Add new method \"update_network\" to \"L3AgentExtension\".","message":"Add new method \"update_network\" to \"L3AgentExtension\".\n\nThis new method will extend the L3 agent extensions functionality,\nmaking them aware of any network change.\n\nChange-Id: I7184c058ea59a1dceefda7ce4488df4d64c71111\nRelated-Bug: #1950454\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4807bacab4ad6320cbd6b3c2c164bad78f5f42d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4807bacab4ad6320cbd6b3c2c164bad78f5f42d5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
