)]}'
{"id":"openstack%2Fneutron-lib~819286","triplet_id":"openstack%2Fneutron-lib~master~I5495568854248dbeba85534ba250fd9b7a607b0c","project":"openstack/neutron-lib","branch":"master","topic":"zuul_update_yoga_testing","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-12-07 14:59:55.000000000","reason":"Change was abandoned"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-12-07 14:59:55.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2021-11-29 13:11:25.000000000","reason":"Lajos Katona replied on the change"}},"hashtags":[],"change_id":"I5495568854248dbeba85534ba250fd9b7a607b0c","subject":"Update setup.cfg for updated testing runtime","status":"ABANDONED","created":"2021-11-25 10:33:36.000000000","updated":"2021-12-07 14:59:55.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f07f159d02d65b26b9a08dc76221427201e6c04d","_number":819286,"virtual_id_number":819286,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-25 11:51:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"date":"2021-11-27 11:45:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2021-11-25 21:19:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":-1,"date":"2021-11-30 09:37:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 10:56:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-11-25 11:51:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 21:19:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"91fcd76d4eb88af807422e319c16db47e0012779","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-25 10:33:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4df9d38a1a42a8e8639f2103456fddd5e55bb201","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-25 10:34:32.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"78cc1867f4c9e5e97efdd0b024e9b1d40c9a3e6b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-25 10:56:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"80d371885845d908ca5ab2adfa7ee63711bc3e92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 11:51:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6c261cb2d1b04fb99e39ca786a751f2f : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6dce788e3f4ba8ba116bd20aed0b92 : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9f135a31e584867a98661442870f4fb : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/874f560378f148b89fddbe96248249d5 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ab7031b60eb4828bfcd29cf4a1248ab : SUCCESS in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e8f2b4533594e93a92ba648f5ca4da2 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39ee9a26f936409ab1fbd4dcc61202cb : SUCCESS in 7m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8397f56a1213419eae6d3169813ccae1 : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f9e8d70c235e1abcecacdd506b7b44092f32ab97","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-25 21:19:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b83b5ec3169b7a1e1df0a0bb61d2f5aa35d24b3","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-27 11:43:54.000000000","message":"Removed Code-Review+2 by Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"dd706fde4216df47d6fc0457c6f70cf1853f6da1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-27 11:45:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c65faf2393d69b3e6f19a5618de67bd674758be9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-29 13:11:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0cd72e6e62d043eb49196c579edb62a38f17c4a1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-30 09:37:27.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"f07f159d02d65b26b9a08dc76221427201e6c04d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-12-07 14:59:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3caac40ae1a4544bc19b4f43dc8bc24b9b31a614","revisions":{"61d735dbdb9067e4d5c5388411db7183d44d1c4d":{"kind":"REWORK","_number":1,"created":"2021-11-25 10:33:36.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/86/819286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/86/819286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/819286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/819286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/819286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/86/819286/1"}}},"commit":{"parents":[{"commit":"0b134560635d48730c8f6dc362ab62645fec2540","subject":"Merge \"Rehome QoS floating IP extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0b134560635d48730c8f6dc362ab62645fec2540"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-11-25 10:32:06.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-11-25 10:32:06.000000000","tz":60},"subject":"Update setup.cfg for updated testing runtime","message":"Update setup.cfg for updated testing runtime\n\nTC has updated the Yoga testing runtime by making the python3.8 and\npython3.9 as python version to tests:\n- https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I5495568854248dbeba85534ba250fd9b7a607b0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/61d735dbdb9067e4d5c5388411db7183d44d1c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/61d735dbdb9067e4d5c5388411db7183d44d1c4d"}]},"branch":"refs/heads/master"},"3caac40ae1a4544bc19b4f43dc8bc24b9b31a614":{"kind":"REWORK","_number":2,"created":"2021-11-25 10:34:32.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/86/819286/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/86/819286/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/819286/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/819286/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/819286/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/86/819286/2"}}},"commit":{"parents":[{"commit":"0b134560635d48730c8f6dc362ab62645fec2540","subject":"Merge \"Rehome QoS floating IP extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0b134560635d48730c8f6dc362ab62645fec2540"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-11-25 10:32:06.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-11-25 10:34:31.000000000","tz":0},"subject":"Update setup.cfg for updated testing runtime","message":"Update setup.cfg for updated testing runtime\n\nTC has updated the Yoga testing runtime by making the python3.8 and\npython3.9 as python version to tests:\n- https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I5495568854248dbeba85534ba250fd9b7a607b0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3caac40ae1a4544bc19b4f43dc8bc24b9b31a614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3caac40ae1a4544bc19b4f43dc8bc24b9b31a614"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
