)]}'
{"id":"openstack%2Fneutron-lib~821806","triplet_id":"openstack%2Fneutron-lib~master~I264cda4ac8d5a737e172a62fa2761fd14fc746f3","project":"openstack/neutron-lib","branch":"master","topic":"bug/1953716","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2021-12-15 11:54:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I264cda4ac8d5a737e172a62fa2761fd14fc746f3","subject":"Allow to pass custom reason in NetworkInUse exception","status":"MERGED","created":"2021-12-15 08:51:24.000000000","updated":"2021-12-15 11:56:15.000000000","submitted":"2021-12-15 11:54:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"821806-bug/1953716","meta_rev_id":"8c70fcc60e384ae56b49a13d289b80fe5a2fb4ed","_number":821806,"virtual_id_number":821806,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-15 11:54:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-12-15 09:14:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-12-15 09:27:36.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-12-15 10:35:07.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-15 09:14:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-12-15 09:27:36.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-12-15 10:28:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"226d8131c45cd11d4db2588b205444a2027203c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-15 08:51:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2874f81c180ed7060317989ee81227eea631b25a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-15 08:58:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9a83f10779cf6fe46c4686acafe88d7c0abf6f91","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-15 09:14:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cd2ccdbd16d49c69ceed802e5e511db6b5ee1ee2","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-12-15 09:27:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"05005daead5861f8acc10b643c4bedab19892ebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 10:28:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0593979986ee4af198f9bfc71b4ca2ab : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1997836674584fb39ce9c7fdc2373eb0 : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4ebc56b07774aca8c496f46c23e69d8 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcff8db503ea4d9787e3069551124dcc : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b65aebad26714618b88a682bb0c36240 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ede867cc187e4b889728ccf8fdffd9b5 : SUCCESS in 3m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44bcc19c1a0b4ba3bfbfd152604cc042 : SUCCESS in 1h 24m 10s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a777b463044b444e9508ee28485250e1 : SUCCESS in 44m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/291252449e804ea1b2c0087dd8213c0a : SUCCESS in 48m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5206f507e83b13d3c41ad903496af3b585d09d9e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-12-15 10:35:07.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dacb46b2bfb832908efb53aa651aae6dc0140ed7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 10:35:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7eca8cf75fef28b48fe2b0c4ae7555d8742ea66a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 11:54:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a9d812835d64427a2a8272714d458fc : SUCCESS in 8m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6ce22a8cf414709b6b9fde651e21ced : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b8a4f31540349319adf25d7211b8659 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25c738c74e9045b4badaf9306abaa224 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa313d037acb4889a45fbdbc992bf4d1 : SUCCESS in 5m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d9d8a9f88bf4ae0b318ef34b4e9462e : SUCCESS in 1h 17m 26s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/97eb692d63894be19510a38133872a1d : SUCCESS in 34m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/174a0cd586b441deb9dd90780590f9f1 : SUCCESS in 40m 51s","accounts_in_message":[],"_revision_number":2},{"id":"c80547783e87200273054dc32e9f44b1fe88cf64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 11:54:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8c70fcc60e384ae56b49a13d289b80fe5a2fb4ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 11:56:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b980e50610d947e8b220f497920aa59c : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a922e45b0be869a22f43609178e77ba0252f6ea","revisions":{"080b513f3944c0ce94569933a9a91e7e42633406":{"kind":"REWORK","_number":1,"created":"2021-12-15 08:51:24.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/06/821806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/06/821806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/06/821806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/06/821806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/06/821806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/06/821806/1"}}},"commit":{"parents":[{"commit":"48345c6e3c923eee5f499e352b2c25aae73e11bf","subject":"Merge \"Replace \"target_tenant\" with \"target_project\" in RBAC OVOs and models\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48345c6e3c923eee5f499e352b2c25aae73e11bf"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-15 08:39:47.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-15 08:48:35.000000000","tz":330},"subject":"Allow to pass custom reason in NetworkInUse exception","message":"Allow to pass custom reason in NetworkInUse exception\n\nIn order to pass custom reason like what ports are\nbeing used on the network, need to allow the Exception\nto except \"reason\" argument.\n\nRelated-Bug: #1953716\nChange-Id: I264cda4ac8d5a737e172a62fa2761fd14fc746f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/080b513f3944c0ce94569933a9a91e7e42633406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/080b513f3944c0ce94569933a9a91e7e42633406"}]},"branch":"refs/heads/master"},"6a922e45b0be869a22f43609178e77ba0252f6ea":{"kind":"REWORK","_number":2,"created":"2021-12-15 08:58:27.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/06/821806/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/06/821806/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/06/821806/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/06/821806/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/06/821806/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/06/821806/2"}}},"commit":{"parents":[{"commit":"48345c6e3c923eee5f499e352b2c25aae73e11bf","subject":"Merge \"Replace \"target_tenant\" with \"target_project\" in RBAC OVOs and models\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48345c6e3c923eee5f499e352b2c25aae73e11bf"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-15 08:39:47.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-15 08:58:09.000000000","tz":330},"subject":"Allow to pass custom reason in NetworkInUse exception","message":"Allow to pass custom reason in NetworkInUse exception\n\nIn order to pass custom reason like what ports are\nbeing used on the network, need to allow the Exception\nto except \"reason\" argument.\n\nRelated-Bug: #1953716\nChange-Id: I264cda4ac8d5a737e172a62fa2761fd14fc746f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6a922e45b0be869a22f43609178e77ba0252f6ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6a922e45b0be869a22f43609178e77ba0252f6ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
