)]}'
{"id":"openstack%2Fneutron-lib~822817","triplet_id":"openstack%2Fneutron-lib~master~Ia741d4d87d3a284add2f2c48f780d5332540fafe","project":"openstack/neutron-lib","branch":"master","topic":"bug/1953478","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-12-23 18:30:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia741d4d87d3a284add2f2c48f780d5332540fafe","subject":"Log resource id in the callbacks publish method","status":"MERGED","created":"2021-12-23 10:57:14.000000000","updated":"2021-12-23 18:31:48.000000000","submitted":"2021-12-23 18:30:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822817-bug/1953478","meta_rev_id":"937e8b493b256ae12ef8eb744da1b048a144eb99","_number":822817,"virtual_id_number":822817,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-23 18:30:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":2,"date":"2021-12-23 17:11:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-12-23 11:08:26.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2021-12-23 17:11:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-23 11:08:26.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-12-23 12:29:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-23 17:11:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"2661455a89f24aa0470c73922ac8c1a1d5a607ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-23 10:57:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1635baa8c7b84b3b45dad0d6de2ee892bf5b079e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-12-23 11:08:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e53dae856c8d7a579b2c68621e9f9772badf1f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 12:29:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d8ddfb1c6a54e5cafc42dc7dfa520b3 : SUCCESS in 5m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27a8ddd984e04fbfbc41916bd115f3c2 : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fe15ddcf5cd474ea55b279f32826c0c : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b938c43ef92466aaaa4dbb1750bfdab : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cf00da2341b4c5fa3260c85fcf10769 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b96d79097f024fb08acf71f5e7afa4a3 : SUCCESS in 4m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c200d401eae42ff8064874d14908d4b : SUCCESS in 1h 27m 10s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5d6d7c8a40f048b7ac78b2d06bc10bad : SUCCESS in 50m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5553ce031d534cc38c4e47407611f2c0 : SUCCESS in 49m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7a016b1deb1f259a22835b37c62c4c607ce34f08","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-12-23 17:11:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"613b4a0cd42929a7eb6451880bb85cd2517cb658","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 17:12:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fbcc1c99e3c479575440cc038aec532157de9827","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 18:30:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad0266e0374e4092bfc25d854afb9836 : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/525ee2cb9247426c99f66c3c2c984926 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7a1fbe3ce81411c86d173623977b2d6 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6608f1b748854c69a92a228798bcd3be : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44be333b2cb94304ab39d4243c8fc9c1 : SUCCESS in 4m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01eec461a14e45bfbf2eb4b1952495b7 : SUCCESS in 1h 17m 00s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/741b4a605a064e539907361906ea60fe : SUCCESS in 39m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/471933614a6e4c658ee035c9d5c181eb : SUCCESS in 42m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a0ba2c43281967c853a31a433eec23203381df2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 18:30:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"937e8b493b256ae12ef8eb744da1b048a144eb99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 18:31:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a61629d475438b82a843ec1cacfdc1 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76b9a095918c02fa90e1cc6f1a481b3a40938b3b","revisions":{"76b9a095918c02fa90e1cc6f1a481b3a40938b3b":{"kind":"REWORK","_number":1,"created":"2021-12-23 10:57:14.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/17/822817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/17/822817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/17/822817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/17/822817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/17/822817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/17/822817/1"}}},"commit":{"parents":[{"commit":"cb031937ed6346dfee6ac3920edfa535cc29a839","subject":"Remove import statement for six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cb031937ed6346dfee6ac3920edfa535cc29a839"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-23 10:55:40.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-23 10:57:10.000000000","tz":60},"subject":"Log resource id in the callbacks publish method","message":"Log resource id in the callbacks publish method\n\nIt may help in debugging if we will know for what resource exactly\ncallbacks are published.\n\nRelated-Bug: #1953478\nChange-Id: Ia741d4d87d3a284add2f2c48f780d5332540fafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/76b9a095918c02fa90e1cc6f1a481b3a40938b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/76b9a095918c02fa90e1cc6f1a481b3a40938b3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
