)]}'
{"id":"openstack%2Fneutron-lib~823857","triplet_id":"openstack%2Fneutron-lib~master~Iad627bcfb95699febd504d7dc7fc9e8a7d813a9e","project":"openstack/neutron-lib","branch":"master","topic":"bug/1956785","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2022-01-11 11:03:17.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-01-11 11:03:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad627bcfb95699febd504d7dc7fc9e8a7d813a9e","subject":"Fix collect duplicates TypeError for dict values","status":"MERGED","created":"2022-01-07 18:49:22.000000000","updated":"2022-01-11 11:05:41.000000000","submitted":"2022-01-11 11:03:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"823857-bug/1956785","meta_rev_id":"ffb023e8aa8fd9ba63acc45350cd45e389596084","_number":823857,"virtual_id_number":823857,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-11 11:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-01-10 14:29:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-01-11 09:31:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-01-11 07:53:22.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-01-10 14:29:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-07 20:28:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 10:32:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-01-10 13:01:01.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-01-10 14:29:48.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-01-11 07:53:22.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-01-11 09:31:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"d6952a048df1d5f4d47934f4a47998a563b41a73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-01-07 18:49:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09cb50fe83c61c9e46d11fd079320478974a281c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 20:28:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a11e06973134653870a7e1f63900b74 : SUCCESS in 7m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ec10bf009924152bc3a18da7a2ffb99 : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/047fca20a99c430492caa3a1fefe1c35 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dd279675a69436b8c1dfce1fdc0c2c8 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58a069ff94804754b5a09ff3505878d4 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b76b0c9e02174bcd84421cfd4c45b4df : SUCCESS in 4m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2784a48a20674c44ae3200a95ff5ed18 : SUCCESS in 1h 36m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c33af18489654752b994d8db1c5d0e2f : SUCCESS in 3m 55s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8cb9abc8fa63498087033a2bd227808f : SUCCESS in 40m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/500cc5d1e926499ca5fe8a6dba58b403 : SUCCESS in 46m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a8952cf2b8c6ac5d2cfc00e4433b8da8dbf9d518","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-10 10:32:24.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"12e31a7ba9fa6f1d5e8d4a6a62685c615e174106","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-10 13:01:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"26b0123299c5e5643732e958d617cefd5491c974","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-10 14:29:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"97938df7d43a705902b3f43c6faf73064d03952c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-10 14:29:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c389777f056cc17079b3ab2472fcfb360f441f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 16:09:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52ececd8ad644ea2b0aad486dbf48f23 : SUCCESS in 9m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bed2369f766645a383e8356e6904ff95 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b866ba24af4040f287e7d0b3678e3be3 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a32304dd292e4eefaf55bfcb64db30f3 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd12e02f9c4c408f8913a5d95bee9ba5 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/324bdeadce094c879041f69348b2f695 : SUCCESS in 6m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56783cffbee3495680f49c82f5fd4983 : SUCCESS in 1h 31m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72d9b6fc458f41e6bc291d52a8ea354b : SUCCESS in 5m 38s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d5b337b7737040b3a434931f912d2229 : SUCCESS in 50m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7d553f176734c0e9fb45b7b923bb374 : SUCCESS in 47m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b29d74bc91c800cada8062519e94e6205d5d2e5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 16:10:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bfae18118b2070307bbdc7cf7e2a4fced214a63f","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-01-10 17:15:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c87e198253de4935d66724d0352d2ffc8efb5a95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 17:43:08.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a333c669b7946fda45227d07531baf9 : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/222e8808c689431bb695355bb73c1cef : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0062128e4af94e4696c25d323932baf0 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d62d4d222fc4a0fad3b54606849e285 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4e5d8ab77aa4bcba8cdb65cb05d7189 : SUCCESS in 4m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5119c76da19a468eadf602f00eff8793 : SUCCESS in 1h 28m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcdd8fa2a8eb4189b46f00e8d5fe9bc5 : SUCCESS in 4m 19s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/529bc1e4f7c84b4487d93f2b97c6c87e : SUCCESS in 42m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8eff0beb9c2c4a009fd63e2404f8aba2 : FAILURE in 51m 22s","accounts_in_message":[],"_revision_number":2},{"id":"42eee0d784ab2a16d70d52abec0de8b580edb8bc","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-01-10 17:45:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c26b2298d39e049cf9002ffee89a64a9ed6e1df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 19:08:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f38c61bf28a74f109982b207e338476e : SUCCESS in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cc189ed21f64d0a8c69bb9e469bb494 : SUCCESS in 6m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1383e00627ce4951833fa3239a2c901d : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/515eac28861643dcb235d3955cd5fbbd : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/598459bd3a4a4c8ea105de9016405f33 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4d0ff1b32ae4f308f61ede3673b4a85 : SUCCESS in 4m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6cf82ff8b9342c395b6a3709def2dd6 : SUCCESS in 1h 20m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77e326297d0742fa8eec9d4696b19e5a : SUCCESS in 4m 11s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ca22dc60faa441bcb28f51f512114917 : SUCCESS in 41m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f59b4fc3708a4361898c31cb43912c61 : SUCCESS in 45m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4ba8e024fab9b943a63c43f2cf6929029c429f94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 19:09:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a68a4a73d2c2908834a135964a83bd6ba2341b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 20:34:38.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6395ae6059684d219bcdeaa6c0b0f6d0 : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a8c46ad1acd452aa9eda08cf713fe0a : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b50d0c2f69c47fb84261db8d1483037 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd20060d253c436f89da2d9784d6791e : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/603c7f93becb4ebdb0820e82a18f0201 : SUCCESS in 4m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b828ac3262e542c587fdc3c3d6966902 : SUCCESS in 1h 21m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2877a80ac6214796b385c0ffe5f0b4d7 : SUCCESS in 4m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e7e4aca2136242bf87f92a97f02e0bc4 : SUCCESS in 44m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3cc2eb0997c547e9b4c0f4cf72148e00 : FAILURE in 50m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b9637bc39e56e1a01bfef4407f182444b488ead8","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-01-11 07:53:22.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d674977c1e40389e13bb3ba32315ba39845c2383","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-11 09:31:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7914724ff9921ea7e2d9dff032ab23e392a97e7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 09:37:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/371aaf1f19d54ca58e462536ae1d4beb : SUCCESS in 5m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f024179fb8dd4448b133d55bd6c521de : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e71db0bd0764c81b9830c958c9e7bb6 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c52ff454cce428ab2735191cfb10ce5 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b99ea15151240b8b7c2c5956e2e2061 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c5e36abbd9244708a95d11dd6289762 : SUCCESS in 5m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c115cba629984b1baf8724069b59c309 : SUCCESS in 1h 40m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae3c1ba220af4d0c97c318629100675a : SUCCESS in 3m 24s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9baee7c9d8dc48b597ff4d88c457c237 : SUCCESS in 45m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9305e8e8cfe54de191c34fd5771d3c7e : SUCCESS in 45m 59s","accounts_in_message":[],"_revision_number":2},{"id":"3d905057690c6175fe78aa518a8c1f97cdef9790","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 09:37:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e4b1dca1dbb4abfdbe14fadadb7dd2e5b4e6b5e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 11:03:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5cd1a7a51c0436298e71dae94650118 : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df46ad0f6c3544d983e8714c221e2dcc : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad02cd240b574a6ab9d9caf76018f10a : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc83a568b6f645e8ae89b08ca2060990 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75b95bd6a92c4f4f93078f994d3cfe9f : SUCCESS in 5m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ed563bbd6704e01b5348f68cb8a2225 : SUCCESS in 1h 24m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f8e0b9f4df24d7a9ebe5489b7ab6503 : SUCCESS in 4m 13s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3be341270e6d47a1bc9e7b76a5695e12 : SUCCESS in 44m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed74184b0f8947e4979e48a052963e51 : SUCCESS in 44m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e6e5c43baad57d270da6803100c1d5fea111d698","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 11:03:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ffb023e8aa8fd9ba63acc45350cd45e389596084","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 11:05:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed656b3a10c644aabe8bd2e4e9bd0560 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4657af735bc648ae8f96f96fa00b1795 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"88b755c0e86414ce8d6b57d6335d22bb892eaffb","revisions":{"aa3901e63143d8fbaafb6ec372b97996eb039a2f":{"kind":"REWORK","_number":1,"created":"2022-01-07 18:49:22.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/57/823857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/823857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/823857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/823857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/823857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/823857/1"}}},"commit":{"parents":[{"commit":"76b9a095918c02fa90e1cc6f1a481b3a40938b3b","subject":"Log resource id in the callbacks publish method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/76b9a095918c02fa90e1cc6f1a481b3a40938b3b"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-01-07 18:21:58.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-01-07 18:49:11.000000000","tz":60},"subject":"Fix collect duplicates TypeError for dict values","message":"Fix collect duplicates TypeError for dict values\n\nThe _collect_duplicates uses a set for duplicates. This\ndoes not work when the values in the list are dictionaries.\n\nInstead of using a set, use a list and extend the condition to\nonly add to the list if seen and not already in dups.\n\nCloses-Bug: #1956785\nChange-Id: Iad627bcfb95699febd504d7dc7fc9e8a7d813a9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/aa3901e63143d8fbaafb6ec372b97996eb039a2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/aa3901e63143d8fbaafb6ec372b97996eb039a2f"}]},"branch":"refs/heads/master"},"88b755c0e86414ce8d6b57d6335d22bb892eaffb":{"kind":"REWORK","_number":2,"created":"2022-01-10 14:29:24.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/57/823857/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/57/823857/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/823857/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/823857/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/57/823857/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/57/823857/2"}}},"commit":{"parents":[{"commit":"76b9a095918c02fa90e1cc6f1a481b3a40938b3b","subject":"Log resource id in the callbacks publish method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/76b9a095918c02fa90e1cc6f1a481b3a40938b3b"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-01-07 18:21:58.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-10 14:29:24.000000000","tz":0},"subject":"Fix collect duplicates TypeError for dict values","message":"Fix collect duplicates TypeError for dict values\n\nThe _collect_duplicates uses a set for duplicates. This\ndoes not work when the values in the list are dictionaries.\n\nInstead of using a set, use a list and extend the condition to\nonly add to the list if seen and not already in dups.\n\nCloses-Bug: #1956785\nChange-Id: Iad627bcfb95699febd504d7dc7fc9e8a7d813a9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/88b755c0e86414ce8d6b57d6335d22bb892eaffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/88b755c0e86414ce8d6b57d6335d22bb892eaffb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
