)]}'
{"id":"openstack%2Fneutron-lib~824926","triplet_id":"openstack%2Fneutron-lib~master~I295a4955bbf1604d4ae2b11adcb5522abe76bc75","project":"openstack/neutron-lib","branch":"master","topic":"taas_api_definitions","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-01-24 08:22:02.000000000","reason":"Slawek Kaplonski replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-01-26 22:42:31.000000000","reason":"Change was submitted"},"33643":{"account":{"_account_id":33643,"name":"Nitish Venkata Patcherla","display_name":"Nitish","email":"nitish.patcherla@est.tech","username":"zitptan"},"last_update":"2022-01-26 22:42:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I295a4955bbf1604d4ae2b11adcb5522abe76bc75","subject":"Move API definitions of taas to neutron-lib","status":"MERGED","created":"2022-01-17 11:24:06.000000000","updated":"2022-01-26 22:44:48.000000000","submitted":"2022-01-26 22:42:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"824926-taas_api_definitions","meta_rev_id":"568f5714da79975255ff6c1882a5ea4d00672be6","_number":824926,"virtual_id_number":824926,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-26 22:42:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2022-01-26 21:15:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-01-24 08:22:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2022-01-26 21:15:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33643,"name":"Nitish Venkata Patcherla","display_name":"Nitish","email":"nitish.patcherla@est.tech","username":"zitptan"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33643,"name":"Nitish Venkata Patcherla","display_name":"Nitish","email":"nitish.patcherla@est.tech","username":"zitptan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-17 11:24:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":33643,"name":"Nitish Venkata Patcherla","display_name":"Nitish","email":"nitish.patcherla@est.tech","username":"zitptan"},"state":"REVIEWER"},{"updated":"2022-01-17 12:12:02.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":33643,"name":"Nitish Venkata Patcherla","display_name":"Nitish","email":"nitish.patcherla@est.tech","username":"zitptan"},"state":"REMOVED"},{"updated":"2022-01-17 12:54:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 07:41:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-01-19 15:44:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":33643,"name":"Nitish Venkata Patcherla","display_name":"Nitish","email":"nitish.patcherla@est.tech","username":"zitptan"},"state":"REVIEWER"},{"updated":"2022-01-26 21:15:15.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"8197ce896ca4cb13920dc4a8828baee0e865a5d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-17 11:24:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"964cd5c5d393b17b56b783c755269db0bed4dc87","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-17 12:12:02.000000000","message":"Removed reviewer Nitish Venkata Patcherla.","accounts_in_message":[],"_revision_number":1},{"id":"81f0e6c4d7134d56866371e471600b6733343385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 12:54:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124f1463670c444b931ceeb6876f1219 : SUCCESS in 5m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7810c26ac6d4509b3f4c85bea93e13c : SUCCESS in 4m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/634892a87de644e3bdb90608ca0953e0 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bf38e539eb54da0875971a4106239d9 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db6f82e2fe8441a9a7381be4c8034dc8 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34270300fc6d42c88a3c76dc4e5e1ef2 : SUCCESS in 5m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/450cc5cd5d2f43519415c5ed95a0d748 : SUCCESS in 1h 25m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c59240e322de4f8390c96b38072d009f : SUCCESS in 3m 59s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c06c9456fd7a4825ac0218a411b619e1 : SUCCESS in 34m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fb4fbb1c1e545fa8844ed8cbbdf9d6b : SUCCESS in 43m 37s","accounts_in_message":[],"_revision_number":1},{"id":"888850dddf3e732a07cc317472cba7f4292ced35","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-18 07:41:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e6994b8790ecdb23978677eb5ddfa2d0a8c4691","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-19 15:44:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8faa97356ebbb592102ca0e71a1772466557a17c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-19 15:46:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d2c2c14e77c41d79d85b9643e2ff42f91e97715","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 17:24:19.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a844a8ec3bdf4758bb6e1203f67aeaaf : SUCCESS in 8m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcf7f086ed4a48cfb601e21df6cc4cd8 : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c564c246065545738a5cd4f848f53b52 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68b31530a46544238043406ecaa497c3 : FAILURE in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c73477cfb6b41a1a1db50580c7b81d1 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4917d7fb2c024ffcad98456a619f8976 : SUCCESS in 6m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dbe1d2e67e3f4149884becac5264b640 : SUCCESS in 1h 35m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c40ee21fb2e4140a8133c09f5517612 : SUCCESS in 5m 19s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/88f45153dae24e51ad4a30389ddf9987 : SUCCESS in 41m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec01af3e5318449b8caa84f05d0c9130 : SUCCESS in 45m 13s","accounts_in_message":[],"_revision_number":2},{"id":"f5acf18237653b698b29dbebdc2968f837656589","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-20 10:07:31.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"641e1f53a49525bcc971975833cbd0fe5fb2542d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 11:30:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af44580e992841f1a06eb45d6e466ce8 : SUCCESS in 7m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1caf5d50fb848dda2d74609d14b8724 : SUCCESS in 7m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cae22adf8d354f98963bd62c2634f2cd : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7561304379e045eb92ab72b4626b0a7d : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/232bfb0dfbbd4c9c9c1ca33aa8328cbd : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a2f51798e5245dbabe93934ee7d4600 : SUCCESS in 6m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f59240158e14a36a8c56adfd08a7fd7 : SUCCESS in 1h 18m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42148d9c01f74749a847b5e7657e63f6 : SUCCESS in 5m 38s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b9736b7f759b4f849082503bc1b4df7d : SUCCESS in 45m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b45595818a44bce9dc30ec35b6fe77b : SUCCESS in 45m 40s","accounts_in_message":[],"_revision_number":3},{"id":"ffd38a006cf5c53095153e056d4af38890d08ee1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-24 08:22:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3f4408dfa151058103b599117898924ffcdf9dbd","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-01-26 21:15:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fa7fa891079df5f12644d3dc1f814bb034694a42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 21:19:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ad0673a535649d55db349b9b2da626fbed7537ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 22:42:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33740cefc2c14caeb9e49ae0d19a904a : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/469276cfb7ed44d5b20ce2403360426b : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0951bf405604425bf763460aaf34508 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdbf33c4a3f549688b9490be74afa848 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06e1e9c8fb6d4af68b7ff602f815ad85 : SUCCESS in 4m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f4be1c10ee94bd2a1c1dec4ad335c30 : SUCCESS in 1h 20m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b2d4674f8dc465287b77b78f8a3b54c : SUCCESS in 5m 03s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3157499fb88f4eae8440326ff96da3d6 : SUCCESS in 42m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/97a59de9ddaf49a89f9b4f56ddc1782e : SUCCESS in 52m 53s","accounts_in_message":[],"_revision_number":3},{"id":"109a2ebf6daf6cc60409b6de1bfc13e4d41654cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 22:42:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"568f5714da79975255ff6c1882a5ea4d00672be6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 22:44:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12dec3692ec840fd8f42baa3bf59c5b4 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad32ba63a4d845d8862df51cab6a137c : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"901589c9a7115cd63351f389ec2f4ffeec02f8fc","revisions":{"9f1508fb5de95e7f21cbc8b430ee21246f4487ea":{"kind":"REWORK","_number":1,"created":"2022-01-17 11:24:06.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/26/824926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/26/824926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/824926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/824926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/824926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/26/824926/1"}}},"commit":{"parents":[{"commit":"8f066c491e0d6a90e80d9aa61fab08e0736a7f30","subject":"Add api-ref for tap service and tap flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8f066c491e0d6a90e80d9aa61fab08e0736a7f30"}]}],"author":{"name":"zitptan","email":"nitish.patcherla@est.tech","date":"2022-01-05 05:50:16.000000000","tz":330},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-01-17 11:23:38.000000000","tz":60},"subject":"Move API definitions of taas to neutron-lib","message":"Move API definitions of taas to neutron-lib\n\nMove tap-as-a-srvice API definitions from taas repo to neutron-lib,\nunder neutron_lib/api/definitions, add necessary unit tests and\nrelease notes.\n\nChange-Id: I295a4955bbf1604d4ae2b11adcb5522abe76bc75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9f1508fb5de95e7f21cbc8b430ee21246f4487ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9f1508fb5de95e7f21cbc8b430ee21246f4487ea"}]},"branch":"refs/heads/master"},"a98e2ed1170d0748268221d6b674ff0ce3f7b8d4":{"kind":"REWORK","_number":2,"created":"2022-01-19 15:44:40.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/26/824926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/26/824926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/824926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/824926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/824926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/26/824926/2"}}},"commit":{"parents":[{"commit":"8f066c491e0d6a90e80d9aa61fab08e0736a7f30","subject":"Add api-ref for tap service and tap flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8f066c491e0d6a90e80d9aa61fab08e0736a7f30"}]}],"author":{"name":"zitptan","email":"nitish.patcherla@est.tech","date":"2022-01-05 05:50:16.000000000","tz":330},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-01-19 15:44:03.000000000","tz":60},"subject":"Move API definitions of taas to neutron-lib","message":"Move API definitions of taas to neutron-lib\n\nMove tap-as-a-srvice API definitions from taas repo to neutron-lib,\nunder neutron_lib/api/definitions, add necessary unit tests and\nrelease notes.\n\nChange-Id: I295a4955bbf1604d4ae2b11adcb5522abe76bc75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a98e2ed1170d0748268221d6b674ff0ce3f7b8d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a98e2ed1170d0748268221d6b674ff0ce3f7b8d4"}]},"branch":"refs/heads/master"},"901589c9a7115cd63351f389ec2f4ffeec02f8fc":{"kind":"REWORK","_number":3,"created":"2022-01-20 10:07:31.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/26/824926/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/26/824926/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/824926/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/824926/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/26/824926/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/26/824926/3"}}},"commit":{"parents":[{"commit":"8f066c491e0d6a90e80d9aa61fab08e0736a7f30","subject":"Add api-ref for tap service and tap flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8f066c491e0d6a90e80d9aa61fab08e0736a7f30"}]}],"author":{"name":"zitptan","email":"nitish.patcherla@est.tech","date":"2022-01-05 05:50:16.000000000","tz":330},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-01-20 10:07:30.000000000","tz":0},"subject":"Move API definitions of taas to neutron-lib","message":"Move API definitions of taas to neutron-lib\n\nMove tap-as-a-srvice API definitions from taas repo to neutron-lib,\nunder neutron_lib/api/definitions, add necessary unit tests and\nrelease notes.\n\nChange-Id: I295a4955bbf1604d4ae2b11adcb5522abe76bc75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/901589c9a7115cd63351f389ec2f4ffeec02f8fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/901589c9a7115cd63351f389ec2f4ffeec02f8fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
