)]}'
{"id":"openstack%2Fneutron-lib~828174","triplet_id":"openstack%2Fneutron-lib~master~I496db96ea40da3bee5b81bcee1edc79e1f46b541","project":"openstack/neutron-lib","branch":"master","topic":"bug/1932154","attention_set":{},"removed_from_attention_set":{"13686":{"account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"last_update":"2022-02-08 20:30:12.000000000","reason":"Change was submitted"},"24824":{"account":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"last_update":"2022-02-08 20:30:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I496db96ea40da3bee5b81bcee1edc79e1f46b541","subject":"Add VNIC_REMOTE_MANAGED for off-path backends","status":"MERGED","created":"2022-02-07 17:36:36.000000000","updated":"2022-02-08 20:51:28.000000000","submitted":"2022-02-08 20:30:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828174-bug/1932154","meta_rev_id":"fb226878cea83f0acd5387b5612b661a164520c9","_number":828174,"virtual_id_number":828174,"owner":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-08 20:30:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2022-02-08 11:46:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2022-02-08 18:22:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-08 07:49:29.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":1,"date":"2022-02-08 11:53:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2022-02-08 18:22:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-07 19:04:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-08 07:49:29.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-02-08 11:46:15.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-02-08 11:53:27.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-02-08 18:22:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"c1b27d5024af37f55d028e9ea4d419479160678f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-07 17:36:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5240c3c4209e7ee49f12320765f5a3ef04037756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 19:04:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/19ce204f3c3c42f38ed19594ce18ec0d : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41869ce74fe243bb8359fda48f3eb02b : SUCCESS in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/160ad12901494692ac02ed8c4db46f2f : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b00eae5942334914b8fa2e889b304dc7 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c267e4e69f214f67beacdae001bf3557 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccacc872a8484904aa737a5673ea5b0d : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96d7440bc042473f9df9c9620a929f29 : SUCCESS in 5m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13b67161604847c99d801ce0fe4a98e7 : SUCCESS in 1h 21m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f9323b19bf6462c9634b10939166649 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dc44ba22c14a4c86a5e7f0c766a63bb4 : SUCCESS in 34m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1519e8dcfdf44ea98f7ccb99d6e36593 : SUCCESS in 45m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a33a820d80e946340350acb55e513610095577b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-08 06:10:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee05d910d7301d7bdb3368fb1ce29e91177b7c74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 07:42:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8c70536714f44c06a3f71cfb0af129b8 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be43cf1c08e340a1b3d0a5cced2dc1e5 : SUCCESS in 6m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca07cd1a78594cef97bad51b2f54d7a4 : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e0518086fcc4487bf072f3a318c257b : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3d3f5f0ca9b4eaeb32d63d7fd7e1bf0 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1114b61400754379bbf5cae8af29e0f4 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e226fd68d8145308ed6a0f530227b92 : SUCCESS in 4m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc543a2f57fe4d75afc2dfd90fbdbdd7 : SUCCESS in 1h 26m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a20dade96ac4174a3e253e97d046159 : SUCCESS in 3m 35s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ef216aa2177745f4b9b867ad9e83a48d : SUCCESS in 1h 11m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4873ffd9dafe4a99844a1e5a5a3280ca : SUCCESS in 53m 35s","accounts_in_message":[],"_revision_number":2},{"id":"f3de531d02f97f039b2a62f0f7852509a5bc8143","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-02-08 07:49:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d0dc58ceebb7d9e1076119e6024f1d17ad2657c5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-02-08 11:46:15.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce06edc12da37c0bce3f62a68004447e7594d21a","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-02-08 11:53:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f553bd630fa9b29375fc99e2773c25b40d3cbe6d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-08 16:37:06.000000000","message":"Topic vnic-remote-managed removed","accounts_in_message":[],"_revision_number":2},{"id":"1568b21ef7cd63d2e322182276fc67ea118b18b6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-08 16:37:10.000000000","message":"Topic set to bug/1932154","accounts_in_message":[],"_revision_number":2},{"id":"e807c7b83dddb319f300232cedd06c89dc884b97","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-02-08 18:22:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2dd1cd7c9b2a83e0ee08918ea743614b7d9a981a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 18:31:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e4f43b4e5cdd62eb3609b61bd0f89c3dc0d59db5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 20:30:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7e4b138cf2bc4d9eb84372b329c1682a : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/982af86ff9b84cd890701b845f7352d4 : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe6233ade1384e4db8670f5cffd04ab1 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa8e8d19dba9406f96f2e74b1f61626b : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9f354e266f54156b5ceddc2c66ad016 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e58bc52e9f514a95992bdd053226c70e : SUCCESS in 5m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/670302f27b6f4cc7988c19a67b18e29c : SUCCESS in 1h 54m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f715b4f22dab42f19e686d4428dc5a8c : SUCCESS in 4m 55s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a218334d2aea4a51b1b747a4fdf7c5e6 : SUCCESS in 34m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bfd683fdf68f4bfa8335a0022b286f50 : SUCCESS in 53m 51s","accounts_in_message":[],"_revision_number":2},{"id":"9bd69b28d82e798196e6eb591e372bf4bb03e0f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 20:30:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fb226878cea83f0acd5387b5612b661a164520c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 20:51:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c384e6cac13440fd88c161f3aa3f8d64 : SUCCESS in 2m 14s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ed81408f2147238f0f949453aa3f4e : SUCCESS in 2m 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29fc244e1fb14fb9a1d497811b0967ca : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"18a69117202080a56a39e32b130e3813c344b9f2","revisions":{"e4b83d01d0a97df5150fce69e890772e1c72398a":{"kind":"REWORK","_number":1,"created":"2022-02-07 17:36:36.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/74/828174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/74/828174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/828174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/828174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/828174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/74/828174/1"}}},"commit":{"parents":[{"commit":"c5fd7066f6dfa34f9825080f6bdb44bae0109055","subject":"Merge \"Add \"connectivity\" property to \"MechanismDriver\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c5fd7066f6dfa34f9825080f6bdb44bae0109055"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-02-07 17:32:58.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-02-07 17:36:34.000000000","tz":180},"subject":"Add VNIC_REMOTE_MANAGED for off-path backends","message":"Add VNIC_REMOTE_MANAGED for off-path backends\n\nAdded a new VNIC type for remote-managed ports of off-path networking\nbackends.\n\nRelated-Bug: #1932154\nChange-Id: I496db96ea40da3bee5b81bcee1edc79e1f46b541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e4b83d01d0a97df5150fce69e890772e1c72398a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e4b83d01d0a97df5150fce69e890772e1c72398a"}]},"branch":"refs/heads/master"},"18a69117202080a56a39e32b130e3813c344b9f2":{"kind":"REWORK","_number":2,"created":"2022-02-08 06:10:58.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/74/828174/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/74/828174/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/828174/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/828174/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/828174/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/74/828174/2"}}},"commit":{"parents":[{"commit":"c5fd7066f6dfa34f9825080f6bdb44bae0109055","subject":"Merge \"Add \"connectivity\" property to \"MechanismDriver\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c5fd7066f6dfa34f9825080f6bdb44bae0109055"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-02-07 17:32:58.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-08 06:10:47.000000000","tz":60},"subject":"Add VNIC_REMOTE_MANAGED for off-path backends","message":"Add VNIC_REMOTE_MANAGED for off-path backends\n\nAdded a new VNIC type for remote-managed ports of off-path networking\nbackends.\n\nRelated-Bug: #1932154\nChange-Id: I496db96ea40da3bee5b81bcee1edc79e1f46b541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/18a69117202080a56a39e32b130e3813c344b9f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/18a69117202080a56a39e32b130e3813c344b9f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
