)]}'
{"id":"openstack%2Fneutron-lib~828276","triplet_id":"openstack%2Fneutron-lib~master~I1a363ec40eb58ec1cd2a9bcdccc4b9a593bcb28d","project":"openstack/neutron-lib","branch":"master","topic":"dh-groups","attention_set":{},"removed_from_attention_set":{"33510":{"account":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"last_update":"2022-10-16 07:12:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1a363ec40eb58ec1cd2a9bcdccc4b9a593bcb28d","subject":"Added new DH groups to vpn to also supports well know EC algotithms.","status":"ABANDONED","created":"2022-02-08 10:50:15.000000000","updated":"2022-10-16 07:12:16.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7b1e4620f5bb1fa678718078ccc0eadb3c659fca","_number":828276,"virtual_id_number":828276,"owner":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-08 08:25:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 12:40:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 12:51:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-04-14 19:35:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"069a5fb064fca88f53eaf3aea53e5ae64e4488a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"date":"2022-02-08 10:50:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a56312d73946c1b89dfa83d38a9c80608e2ce70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 12:40:42.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc6a6b9f1c634b3e8fa1fb608919a02d : SUCCESS in 5m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1258d5bb7f4447ae8e829b44db7ac54f : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/406b6ae1a25e4ff4b671b08fa138232d : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebeb2f69a5b4422c867e9b93b0e78b1a : FAILURE in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/560a3ef80de24de59c87676dd955a4eb : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3be3b26e0bd24f84bd28ec6f26080a9d : SUCCESS in 4m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b77fbe0aba3b4932afe1904d402ff6d4 : SUCCESS in 1h 42m 34s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/600748ace6b2436ca4d31b70e5bb0f51 : SUCCESS in 54m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/70c91da6dca548acb11d62bee85af6da : SUCCESS in 42m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c42043d6bf7e24b10e1b326f3839d3bf56a54a4c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-09 12:51:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fab08e1d427271ce1d7e6ebb384671e5dda6cdd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"date":"2022-02-15 07:54:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"96b22257d4c539949512198341c42fec0eb34f7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 09:22:42.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e73e54dd9f1432b8c21d9ba178f1d1d : SUCCESS in 6m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e2ec0fa56c248619f46eef37dd059be : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56504c276010469491520357b223e66e : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4211df28d9b8418d82302acff82fce64 : FAILURE in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f06f7081604847d9a5bd79ed49855671 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a1d8d19dff84944a0db9b68c77332b8 : SUCCESS in 3m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81dc638d569145b689aa4809b5af24d2 : SUCCESS in 1h 25m 16s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7299bcb8624347e6806d0f9fa5656a0e : SUCCESS in 41m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4ebbf765f3f4edea2328c9438285b12 : SUCCESS in 48m 33s","accounts_in_message":[],"_revision_number":2},{"id":"7faf3a7560257a1da03bac83957f5dfa8bad9b86","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-14 19:35:30.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2},{"id":"dd560a7cd0be3c6a17e765baad02e85a61c98d67","tag":"autogenerated:gerrit:restore","author":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"date":"2022-09-08 08:24:50.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"166e77c71c4692292d7454f0ba09c9182cc930a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"date":"2022-09-08 08:25:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"39667bcb00d740219b7dfafc83c6c38b9814afc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 08:25:33.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron-vpnaas for 828275,3","accounts_in_message":[],"_revision_number":3},{"id":"07bb23a683792e93ac92873c0ae8100e4c300f30","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-16 07:39:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b16d1fbc86e6f445f29fb9be38af431ec94ab6b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 07:39:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2ba7e230fb245af969f52f29a5d53c5\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":3},{"id":"7b1e4620f5bb1fa678718078ccc0eadb3c659fca","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-10-16 07:12:16.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ab3991bff0f42d25cacfbe0bddcbda4c575c66fc","revisions":{"59fbbd19f81e71f73c96deddf7314aa3c6093436":{"kind":"REWORK","_number":1,"created":"2022-02-08 10:50:15.000000000","uploader":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"ref":"refs/changes/76/828276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/76/828276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/828276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/828276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/828276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/76/828276/1"}}},"commit":{"parents":[{"commit":"c5fd7066f6dfa34f9825080f6bdb44bae0109055","subject":"Merge \"Add \"connectivity\" property to \"MechanismDriver\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c5fd7066f6dfa34f9825080f6bdb44bae0109055"}]}],"author":{"name":"Matus Brandys","email":"matus.brandys@vnet.eu","date":"2022-02-08 10:45:30.000000000","tz":60},"committer":{"name":"Matus Brandys","email":"matus.brandys@vnet.eu","date":"2022-02-08 10:50:13.000000000","tz":60},"subject":"Added new DH groups to vpn to also supports well know EC algotithms. Changes also submited to neutron-vpnaas project with commit ID 6c1369f1b3ec44651c041f7c2477d95464865397","message":"Added new DH groups to vpn to also supports well know EC algotithms. Changes also submited to neutron-vpnaas project with commit ID 6c1369f1b3ec44651c041f7c2477d95464865397\n\nChange-Id: I1a363ec40eb58ec1cd2a9bcdccc4b9a593bcb28d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/59fbbd19f81e71f73c96deddf7314aa3c6093436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/59fbbd19f81e71f73c96deddf7314aa3c6093436"}]},"branch":"refs/heads/master"},"4335ed96b473ec14c8d817bd5ae7984153fbc143":{"kind":"REWORK","_number":2,"created":"2022-02-15 07:54:44.000000000","uploader":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"ref":"refs/changes/76/828276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/76/828276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/828276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/828276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/828276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/76/828276/2"}}},"commit":{"parents":[{"commit":"c5fd7066f6dfa34f9825080f6bdb44bae0109055","subject":"Merge \"Add \"connectivity\" property to \"MechanismDriver\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c5fd7066f6dfa34f9825080f6bdb44bae0109055"}]}],"author":{"name":"Matus Brandys","email":"matus.brandys@vnet.eu","date":"2022-02-08 10:45:30.000000000","tz":60},"committer":{"name":"Matus Brandys","email":"matus.brandys@vnet.eu","date":"2022-02-15 07:54:44.000000000","tz":0},"subject":"Added new DH groups to vpn to also supports well know EC algotithms. Changes also submited to neutron-vpnaas project with commit ID 6c1369f1b3ec44651c041f7c2477d95464865397","message":"Added new DH groups to vpn to also supports well know EC algotithms. Changes also submited to neutron-vpnaas project with commit ID 6c1369f1b3ec44651c041f7c2477d95464865397\n\nChange-Id: I1a363ec40eb58ec1cd2a9bcdccc4b9a593bcb28d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4335ed96b473ec14c8d817bd5ae7984153fbc143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4335ed96b473ec14c8d817bd5ae7984153fbc143"}]},"branch":"refs/heads/master"},"ab3991bff0f42d25cacfbe0bddcbda4c575c66fc":{"kind":"REWORK","_number":3,"created":"2022-09-08 08:25:08.000000000","uploader":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"ref":"refs/changes/76/828276/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/76/828276/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/828276/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/828276/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/828276/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/76/828276/3"}}},"commit":{"parents":[{"commit":"c5fd7066f6dfa34f9825080f6bdb44bae0109055","subject":"Merge \"Add \"connectivity\" property to \"MechanismDriver\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c5fd7066f6dfa34f9825080f6bdb44bae0109055"}]}],"author":{"name":"Matus Brandys","email":"matus.brandys@vnet.eu","date":"2022-02-08 10:45:30.000000000","tz":60},"committer":{"name":"Matus Brandys","email":"matus.brandys@vnet.eu","date":"2022-09-08 08:24:03.000000000","tz":120},"subject":"Added new DH groups to vpn to also supports well know EC algotithms.","message":"Added new DH groups to vpn to also supports well know EC algotithms.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-vpnaas/+/828275\nChange-Id: I1a363ec40eb58ec1cd2a9bcdccc4b9a593bcb28d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ab3991bff0f42d25cacfbe0bddcbda4c575c66fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ab3991bff0f42d25cacfbe0bddcbda4c575c66fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
