)]}'
{"id":"openstack%2Fneutron-lib~828487","triplet_id":"openstack%2Fneutron-lib~master~Ie538d611c53ba7bb2fadc98bfd64f1ed0cb9c6bf","project":"openstack/neutron-lib","branch":"master","topic":"dh-groups","attention_set":{},"removed_from_attention_set":{"33510":{"account":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"last_update":"2022-10-16 07:12:56.000000000","reason":"Change was abandoned"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-04-14 19:35:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie538d611c53ba7bb2fadc98bfd64f1ed0cb9c6bf","subject":"Fixed issue with maximum characters per line.","status":"ABANDONED","created":"2022-02-09 08:50:35.000000000","updated":"2022-10-16 07:12:56.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b4ab808b8b740cc305e0edff820e334524bdb1ef","_number":828487,"virtual_id_number":828487,"owner":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-09 10:18:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":-1,"date":"2022-02-09 12:51:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-02-09 09:24:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 09:24:09.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-02-09 10:18:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 12:51:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-04-14 19:35:41.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"e72516d96074bd3815fe01e73f985117e8eb4368","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"date":"2022-02-09 08:50:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d26532d24e3c86c9e24ea7361be5e7614564b29a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-02-09 09:24:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5fa6212d557d62bd4d70bc1a596221e6beb926da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 10:18:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32c1f4c5cd6419e84a8e897c0fa84cb : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8fdcb5aff9f4001883f3c1dd488662e : SUCCESS in 8m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f5c73a1c66c43619806b9f8ddde3196 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cd17e7a82064673ac22f6bc2033f455 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fc1f29326d24c1287f121506f19d05a : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/499ab460ec064a9bade2de4f862e5721 : SUCCESS in 5m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/022aaf78d8804923892d2d1046551164 : FAILURE in 1h 19m 34s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bf155cac9718404e851dccb016e76723 : SUCCESS in 53m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/13daaddb512849b282d882e1c37cbbfa : SUCCESS in 51m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5893a9e36fd337f8be1aca9b621079c3bb21f6a4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-09 12:51:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4f6ff2103f51bfad1314cbc6034364979df151f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-14 19:35:41.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"3bc90b0a4f1a8ac62ba65d19175faf80889dd347","tag":"autogenerated:gerrit:restore","author":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"date":"2022-09-08 08:34:51.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"b7c78b6c074ab275b3b4a49832396554f711823f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 08:35:17.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron-vpnaas for 828275,3","accounts_in_message":[],"_revision_number":1},{"id":"b4ab808b8b740cc305e0edff820e334524bdb1ef","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-10-16 07:12:56.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3ab04d0fff1886589a420d0f7cfdf1fbb073eb4","revisions":{"d3ab04d0fff1886589a420d0f7cfdf1fbb073eb4":{"kind":"REWORK","_number":1,"created":"2022-02-09 08:50:35.000000000","uploader":{"_account_id":33510,"name":"Matus Brandys","email":"matus.brandys@vnet.eu","username":"brando"},"ref":"refs/changes/87/828487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/87/828487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/828487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/828487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/87/828487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/87/828487/1"}}},"commit":{"parents":[{"commit":"59fbbd19f81e71f73c96deddf7314aa3c6093436","subject":"Added new DH groups to vpn to also supports well know EC algotithms. Changes also submited to neutron-vpnaas project with commit ID 6c1369f1b3ec44651c041f7c2477d95464865397","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/59fbbd19f81e71f73c96deddf7314aa3c6093436"}]}],"author":{"name":"Matus Brandys","email":"matus.brandys@vnet.eu","date":"2022-02-09 08:44:08.000000000","tz":60},"committer":{"name":"Matus Brandys","email":"matus.brandys@vnet.eu","date":"2022-02-09 08:44:08.000000000","tz":60},"subject":"Fixed issue with maximum characters per line.","message":"Fixed issue with maximum characters per line.\n\nChange-Id: Ie538d611c53ba7bb2fadc98bfd64f1ed0cb9c6bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d3ab04d0fff1886589a420d0f7cfdf1fbb073eb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d3ab04d0fff1886589a420d0f7cfdf1fbb073eb4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
