)]}'
{"id":"openstack%2Fneutron-lib~832338","triplet_id":"openstack%2Fneutron-lib~master~I0e11885e15616509942d60fbced3ff8e14cac654","project":"openstack/neutron-lib","branch":"master","topic":"upstream/change/add-support-to-valid-ranges-on-port-ranges-validator","attention_set":{},"removed_from_attention_set":{"30695":{"account":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"last_update":"2022-03-16 05:24:37.000000000","reason":"Change was submitted"},"30380":{"account":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"last_update":"2022-03-15 00:11:08.000000000","reason":"ZhouHeng replied on the change"}},"hashtags":[],"change_id":"I0e11885e15616509942d60fbced3ff8e14cac654","subject":"Add custom range parameters on port_range validator","status":"MERGED","created":"2022-03-07 21:14:38.000000000","updated":"2022-03-16 05:25:59.000000000","submitted":"2022-03-16 05:24:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832338-upstream/change/add-support-to-valid-ranges-on-port-ranges-validator","meta_rev_id":"f64fc3e872da0c3d012871ee6a796ce78bc10465","_number":832338,"virtual_id_number":832338,"owner":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-16 05:24:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"all":[{"value":1,"date":"2022-03-15 14:49:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":2,"date":"2022-03-15 15:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-15 00:11:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":1,"date":"2022-03-15 15:40:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-07 23:02:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-08 02:30:35.000000000","updated_by":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"REVIEWER"},{"updated":"2022-03-15 14:49:50.000000000","updated_by":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"reviewer":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"state":"REVIEWER"},{"updated":"2022-03-15 15:40:52.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"da5006ca7820ebdad2a5d1c367acd5a3a371c482","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2022-03-07 21:14:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"026000e9c0fb048c5bd13e0b1154546d94079ce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 23:02:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d58cc2aaf2042498e2fc78e325de446 : SUCCESS in 6m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/150ad01197324b5e9c00701975875f5b : FAILURE in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2bc2fd86e894591b7e4297920e62d2c : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a3c3e85b3b4db0b72bf84484ead9e7 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf3106e065e14029bf2afa92017dc30c : FAILURE in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/104c3122e5b0413da7a23c9f8c61d0aa : FAILURE in 6m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e85e3f572b5347028965bec2bffb1f3f : SUCCESS in 1h 42m 25s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/46fb9ba7b52e46e6a6e2c8301b659b18 : SUCCESS in 47m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/33cb86a0c75d47c191220baa7f2711a6 : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ee106f7ac6e252cf0b3053a28765a49dd4bdc36a","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-08 05:47:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f1667a11bfcccfe13e2fb637d2eb08192d157fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2022-03-08 11:41:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eea54af68336bfc22058da2f35ee660de20d3483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 13:16:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e77bd4103f8a4edaaadf03d955465f59 : SUCCESS in 10m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/108b00b3152144fcba073dc69b1276a6 : FAILURE in 6m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02537d7f3dae43b79255d6b45f321bb8 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afd250136cc14a2d82b1011e34ec2942 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f31aa1e59fff4a8c9155c7f7c68ac790 : FAILURE in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f258fa7067ba437188c3f5245f3fc0e4 : FAILURE in 6m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f705e152e7be4d35b2a629101ca9f27e : SUCCESS in 1h 23m 40s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/12c84afb5f17470e996664906efacbaa : SUCCESS in 41m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ffcf786259c64f6d843ecc6626a3fa95 : SUCCESS in 47m 21s","accounts_in_message":[],"_revision_number":2},{"id":"2fc7f4638487d9a83a242b9701f54999eb92046d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-08 23:43:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"66746e82ecbfdee95812a094b11c605b6adb6eda","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-09 01:27:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"22a125cb5ee31c34a887c05aa0a8ed65eec12f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 01:44:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cabc2895df7d460d9327844760bda500 : SUCCESS in 7m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55e6319f16024ef7a82a7401b6c74111 : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab8bc63fd0ef4ae3a9de19c79d2ed573 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2850a33f7704d198c77b0958fcc4137 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c90c29547bc84d76841a58ad76d055ec : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eddd5f484d4546ecb59df2fddd94707a : SUCCESS in 5m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1165ed47c591492589a9794c36b34e6a : SUCCESS in 1h 52m 47s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b5061abb14224cbe817e67d54977dc29 : SUCCESS in 46m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2ce204ae09c64a359f56f55c3a45753a : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":3},{"id":"6d3a67637757f5301888c572a217fd036efefb8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2022-03-09 21:16:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"360e6c5bee73f494ffa80483171d9e8314e11ce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 22:43:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c14b39e61ef24670acd8c063ce2b7d00 : SUCCESS in 7m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b63879a5edfb46f3a106a3be058f93db : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c149a7ace4004ea8aef4c784d4323c12 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a8c4fd6699243b7aac2a4ba5f4df24f : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cc598cc571a4d828b2309ba683c19de : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/135923413cd44ce58df17184131722c3 : SUCCESS in 5m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10335948dfa94053a83527e468a45772 : SUCCESS in 1h 25m 00s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0935ee134d574e49a65dc3b86cfa93fb : SUCCESS in 54m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/74847e6112ff40fc8a9b2a8e8478ecf4 : SUCCESS in 45m 14s","accounts_in_message":[],"_revision_number":4},{"id":"4a951f9daa065c3018fda295c661fbd47dda276f","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-14 00:32:37.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"98c1675972f15ad254cc589d0473abf27b95a951","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2022-03-14 11:42:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"faf032d00f7ffd093850e5d70dfcccfd98ea209f","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2022-03-14 11:44:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"86433194caf117f981f77ad4fc9ff2dea2eaaaa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 13:22:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6faaa45a33634f6f8b5c10f95ebd699b : SUCCESS in 6m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c46fe3e4133242a9a62b2972d0a5e0be : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d373a3a6f77b4c62882300f66ca76127 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5f8f7de8fa84fb5b970f713fca290e7 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2de764b022a44d30961474a54fed192d : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a814da1cc5ec4326b694c39f3261d065 : SUCCESS in 5m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4635776eabc347d9a0ae090ce6a92534 : SUCCESS in 1h 37m 31s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b0c03c2c2528442c8134e082b530a3ec : SUCCESS in 36m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0377aabaa6134c47af8afa86e08379fb : FAILURE in 49m 31s","accounts_in_message":[],"_revision_number":5},{"id":"402066c3a64e091356da24a4e2a12f190eabf6fc","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2022-03-14 22:26:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8c9f1bcf5b4496c272e48735f3f573ee6e55a1a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 23:49:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84baee2c5f744e33956645b173652d38 : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e503a4cb3de8438e92211d154db1312b : SUCCESS in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d7c4e905b8e404e8dab0a7fc6a7f22c : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6015caa38179405a9b89b41bcce505e6 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0c82a0527414e92b00c2217335e523d : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0103ae177626407789a60334981cb08c : SUCCESS in 5m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc0a50a7d6db470bb767e81648fac2c6 : SUCCESS in 1h 21m 02s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c2c7cdccbb1d45e881a235a7f3950cae : SUCCESS in 40m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/62b9868d57cd4ba0903d44133ddeba5a : SUCCESS in 47m 10s","accounts_in_message":[],"_revision_number":5},{"id":"4ff348e9e9c3aae5f9b35eeb87b2fa90f13879bd","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-15 00:11:08.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"485be3c5797acf07d75f69ebff6dd6b4951e7a9f","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-03-15 14:49:50.000000000","message":"Patch Set 5: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b99ff867ba016a1e604b14408babadb9aa74482e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-15 15:40:52.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"962a4d20d74f5e7c34aba3dd3f2f4eedf3971497","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 15:41:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b80ff4b1387b72767f748bf0d474f1cd0401debb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:14:46.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/132882a1c11d422eba1b00354e877f94 : SUCCESS in 7m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da0d645d6eb1421a8094f5d04b0c725a : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/655b5818692546f89837d2891a342e30 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7da3a6f26b649feb02c3cc5d4ca6d71 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ce65f17fc824d4b92d5a9cde343734a : SUCCESS in 6m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d3953a2d7704666ab06f13f13b0e342 : SUCCESS in 1h 31m 56s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c7b91204e42048f29affcd50bc9d8ad3 : SUCCESS in 47m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ddba41b2f12a4128a105430554053494 : FAILURE in 49m 42s","accounts_in_message":[],"_revision_number":5},{"id":"256fbb5a1bd42cb7ef7f1ed482c1cf45bf87a4b6","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-03-15 17:15:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"02c4c88734f726fd0a109936509676db8ff70d8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 18:46:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1ce56c71654c82aea703984d846814 : SUCCESS in 7m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f759d15f39dc4f35a981f62dbfc4f125 : SUCCESS in 5m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa019db0623b4e4db24b29f9d356e0a9 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54f8153a38834db0843eba81f67b15c6 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35ba144487704fe8a6c860202da5185e : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d3e75969f6a4b2f957e6a729e554e34 : SUCCESS in 5m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f025d51e9284e678f29e5b1e153d604 : SUCCESS in 1h 25m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dd7760890a6842bf848445603c1cf5e4 : SUCCESS in 48m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/40c47e03df6041e887f9555ff0c58104 : SUCCESS in 52m 54s","accounts_in_message":[],"_revision_number":5},{"id":"e8961302d9374c63b5f79f816e4dd0746224db33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 18:47:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5a6c7f189bd884393e67c811eb2151d27dbc07ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 20:20:42.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62cd8cd4177d40de988eb71e6ca5db4e : SUCCESS in 7m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a82c1ab43d245e7a8d8b2b897ff611a : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff0d5146dfe44767be83a491c7703cd7 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fc795098cc14ef39e64be1a53cb4c67 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/644e3738a56e4790a30f92e126cbba3b : SUCCESS in 5m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4dcfe764af445a8b4a6abfa37785f1c : SUCCESS in 1h 29m 35s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5a5b51713b6941b9b62a6ad8da33c71c : SUCCESS in 47m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a6aafb9163ee40e384d9214b02e47af3 : FAILURE in 56m 43s","accounts_in_message":[],"_revision_number":5},{"id":"ae2d1e0e420168cd8974f144a58fc2755ba56627","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-03-15 20:22:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"41c7359ebe8e0bc54fbebcbed5227f04b655d794","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 21:43:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05bc450bf654264a76e6753ae5873f9 : SUCCESS in 6m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dd6df55c2cc4d99aba177f37cfecaaa : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ba248ef341045a485c1d12feb323c48 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a359652ef4a4b1095c08a3299c95a3d : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/752eb2acf5a0436d9ece7f8b426174b3 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7bee01202ca4a89ac2c603210aa1021 : SUCCESS in 5m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/754389fa4e174bbda7efd1190fb95ab3 : SUCCESS in 1h 17m 53s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b248d0aa76414dfc8108bede2afdd118 : SUCCESS in 31m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2dbc45ad463c4cfab681cd16ffde063b : FAILURE in 42m 04s","accounts_in_message":[],"_revision_number":5},{"id":"5f2f764d96ba26e259a08b2961037508ebd5b0ab","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-15 23:43:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c819e80a62ee03f030a8c538f45a1b9fdce6d42b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 01:19:55.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b57f460824ec4d92a6ace77bf4984e80 : SUCCESS in 6m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea9b897d6c444fa5a968983afb9f8867 : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/477a2c172b3c4d47a6123d0159e7b5cc : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2db84657cec4b61882f96dd500ef71f : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08f5b135d6ae4329a25167bebd347ec3 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b18d872d7a284926879b7c7862e63aa5 : SUCCESS in 5m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7949eda938b2481c953e38e9d66700df : SUCCESS in 1h 31m 52s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/70e042d827c1496089c07e13f50adabf : SUCCESS in 51m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e3e114d7909436c93edcb78a5ea6ea6 : FAILURE in 52m 46s","accounts_in_message":[],"_revision_number":5},{"id":"7262485255d4bd367475424f0b50590fb97a40c4","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-16 02:21:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0f990cc47c86196c9e965135ea0aa3eea49ca75a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 03:55:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b68c9ad5d59447e836e88cac7989f53 : SUCCESS in 6m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22151c61c10847a586e5e102b41c2681 : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70ea77f853be44f697d3d4ca5555892f : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e468d8a60da4b54b67b68fc32f84301 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01ababc677e94697a077a43ad5b1a797 : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9468fc33ce2439b89a301a3a04da6e1 : SUCCESS in 8m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6776981f3b5c4b13bd85150ec9683d92 : SUCCESS in 1h 24m 15s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/eb72ff5585834e4b9550d8b2ef1001dc : SUCCESS in 48m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a58555768a94bde9f80100a87b54502 : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":5},{"id":"0d944192695f18f005359dfe3a3a3c1a3d3c4256","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 03:56:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bc427aaf0e08b034de390aa7a83113373146aed5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 05:24:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c167f22d2f145fdacdfb420dd5a392b : SUCCESS in 7m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b387240cabba48dca64d44449433544a : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed5f081ad46949dd9d94b4f4e521ba87 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf9bfbca680e4c0896dd7e3ff798d394 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cf03f91e08e405faa00bfdf1daffd49 : SUCCESS in 6m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b72e2c68aadb4a55ae5ba7942e647e19 : SUCCESS in 1h 26m 48s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fa4e76f69790410e8c2333564aebb597 : SUCCESS in 43m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/127a4094dfb449dda4c873adae19409a : SUCCESS in 53m 37s","accounts_in_message":[],"_revision_number":5},{"id":"f1d9d11affeae7d48532b3c6d1344c2a7cd29a8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 05:24:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"f64fc3e872da0c3d012871ee6a796ce78bc10465","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 05:25:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d63263a10f45ff8f31a6a86f3969c0 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1dde6c24b777a86eb1a562a4ee2ca5019ac1e08a","revisions":{"3467868eb4cf201dc78ff6203e0ba3d1a7e7f93a":{"kind":"REWORK","_number":1,"created":"2022-03-07 21:14:38.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/38/832338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/832338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/1"}}},"commit":{"parents":[{"commit":"f765e3c21c52f97c713697db4d84c526054078f3","subject":"Rehome exception ``L3ExtensionException``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f765e3c21c52f97c713697db4d84c526054078f3"}]}],"author":{"name":"Pedro Henrique","email":"phpm13@gmail.com","date":"2022-03-07 14:56:18.000000000","tz":-180},"committer":{"name":"Pedro Henrique","email":"phpm13@gmail.com","date":"2022-03-07 14:56:18.000000000","tz":-180},"subject":"Add custom range parameters on port_range validator","message":"Add custom range parameters on port_range validator\n\nTo maintain the behavior of port forwarding\nvalidations that denies the port number 0 when\ncreating a new port forwarding rule for floating\nIPs, we will need to extend the\n\u0027validate_port_range_or_none\u0027 validator to allow\nus to define the acceptable port ranges when\nvalidating a value.\n\nThis patch is one of a series of patches\nto implement floating ip port forwarding with\nport ranges.\n\nThe specification is defined in:\nhttps://github.com/openstack/neutron-specs/blob/master/specs/wallaby/port-forwarding-port-ranges.rst\n\nChange-Id: I0e11885e15616509942d60fbced3ff8e14cac654\nImplements: blueprint floatingips-portforwarding-ranges\nRelated-Bug: #1885921\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3467868eb4cf201dc78ff6203e0ba3d1a7e7f93a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3467868eb4cf201dc78ff6203e0ba3d1a7e7f93a"}]},"branch":"refs/heads/master"},"5fd9258e8005de814ae9b020aa9279230434185d":{"kind":"REWORK","_number":2,"created":"2022-03-08 11:41:45.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/38/832338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/832338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/2"}}},"commit":{"parents":[{"commit":"f765e3c21c52f97c713697db4d84c526054078f3","subject":"Rehome exception ``L3ExtensionException``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f765e3c21c52f97c713697db4d84c526054078f3"}]}],"author":{"name":"Pedro Henrique","email":"phpm13@gmail.com","date":"2022-03-07 14:56:18.000000000","tz":-180},"committer":{"name":"Pedro Henrique","email":"phpm13@gmail.com","date":"2022-03-08 11:41:05.000000000","tz":-180},"subject":"Add custom range parameters on port_range validator","message":"Add custom range parameters on port_range validator\n\nTo maintain the behavior of port forwarding\nvalidations that denies the port number 0 when\ncreating a new port forwarding rule for floating\nIPs, we will need to extend the\n\u0027validate_port_range_or_none\u0027 validator to allow\nus to define the acceptable port ranges when\nvalidating a value.\n\nThis patch is one of a series of patches\nto implement floating ip port forwarding with\nport ranges.\n\nThe specification is defined in:\nhttps://github.com/openstack/neutron-specs/blob/master/specs/wallaby/port-forwarding-port-ranges.rst\n\nChange-Id: I0e11885e15616509942d60fbced3ff8e14cac654\nImplements: blueprint floatingips-portforwarding-ranges\nRelated-Bug: #1885921\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5fd9258e8005de814ae9b020aa9279230434185d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5fd9258e8005de814ae9b020aa9279230434185d"}]},"branch":"refs/heads/master"},"d11460939f7ad20c1f44749c5c5a5f076bbd41fc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-03-08 23:43:52.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/38/832338/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/832338/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/3"}}},"commit":{"parents":[{"commit":"a8abe9d592da5bcf065af40e0ba1cd3599ede1e7","subject":"[test]Remove the validation of the tenant property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a8abe9d592da5bcf065af40e0ba1cd3599ede1e7"}]}],"author":{"name":"Pedro Henrique","email":"phpm13@gmail.com","date":"2022-03-07 14:56:18.000000000","tz":-180},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2022-03-08 23:43:52.000000000","tz":0},"subject":"Add custom range parameters on port_range validator","message":"Add custom range parameters on port_range validator\n\nTo maintain the behavior of port forwarding\nvalidations that denies the port number 0 when\ncreating a new port forwarding rule for floating\nIPs, we will need to extend the\n\u0027validate_port_range_or_none\u0027 validator to allow\nus to define the acceptable port ranges when\nvalidating a value.\n\nThis patch is one of a series of patches\nto implement floating ip port forwarding with\nport ranges.\n\nThe specification is defined in:\nhttps://github.com/openstack/neutron-specs/blob/master/specs/wallaby/port-forwarding-port-ranges.rst\n\nChange-Id: I0e11885e15616509942d60fbced3ff8e14cac654\nImplements: blueprint floatingips-portforwarding-ranges\nRelated-Bug: #1885921\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d11460939f7ad20c1f44749c5c5a5f076bbd41fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d11460939f7ad20c1f44749c5c5a5f076bbd41fc"}]},"branch":"refs/heads/master","description":"Rebase"},"24272da190ec5fbb694d603e1f51802ddcc0bc62":{"kind":"REWORK","_number":4,"created":"2022-03-09 21:16:12.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/38/832338/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/832338/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/4"}}},"commit":{"parents":[{"commit":"f765e3c21c52f97c713697db4d84c526054078f3","subject":"Rehome exception ``L3ExtensionException``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f765e3c21c52f97c713697db4d84c526054078f3"}]}],"author":{"name":"Pedro Henrique","email":"phpm13@gmail.com","date":"2022-03-07 14:56:18.000000000","tz":-180},"committer":{"name":"Pedro Henrique","email":"phpm13@gmail.com","date":"2022-03-09 21:15:57.000000000","tz":-180},"subject":"Add custom range parameters on port_range validator","message":"Add custom range parameters on port_range validator\n\nTo maintain the behavior of port forwarding\nvalidations that denies the port number 0 when\ncreating a new port forwarding rule for floating\nIPs, we will need to extend the\n\u0027validate_port_range_or_none\u0027 validator to allow\nus to define the acceptable port ranges when\nvalidating a value.\n\nThis patch is one of a series of patches\nto implement floating ip port forwarding with\nport ranges.\n\nThe specification is defined in:\nhttps://github.com/openstack/neutron-specs/blob/master/specs/wallaby/port-forwarding-port-ranges.rst\n\nChange-Id: I0e11885e15616509942d60fbced3ff8e14cac654\nImplements: blueprint floatingips-portforwarding-ranges\nRelated-Bug: #1885921\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/24272da190ec5fbb694d603e1f51802ddcc0bc62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/24272da190ec5fbb694d603e1f51802ddcc0bc62"}]},"branch":"refs/heads/master"},"1dde6c24b777a86eb1a562a4ee2ca5019ac1e08a":{"kind":"REWORK","_number":5,"created":"2022-03-14 11:42:49.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/38/832338/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/832338/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/832338/5"}}},"commit":{"parents":[{"commit":"f765e3c21c52f97c713697db4d84c526054078f3","subject":"Rehome exception ``L3ExtensionException``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f765e3c21c52f97c713697db4d84c526054078f3"}]}],"author":{"name":"Pedro Henrique","email":"phpm13@gmail.com","date":"2022-03-07 14:56:18.000000000","tz":-180},"committer":{"name":"Pedro Henrique","email":"phpm13@gmail.com","date":"2022-03-14 11:42:21.000000000","tz":-180},"subject":"Add custom range parameters on port_range validator","message":"Add custom range parameters on port_range validator\n\nTo maintain the behavior of port forwarding\nvalidations that denies the port number 0 when\ncreating a new port forwarding rule for floating\nIPs, we will need to extend the\n\u0027validate_port_range_or_none\u0027 validator to allow\nus to define the acceptable port ranges when\nvalidating a value.\n\nThis patch is one of a series of patches\nto implement floating ip port forwarding with\nport ranges.\n\nThe specification is defined in:\nhttps://github.com/openstack/neutron-specs/blob/master/specs/wallaby/port-forwarding-port-ranges.rst\n\nChange-Id: I0e11885e15616509942d60fbced3ff8e14cac654\nImplements: blueprint floatingips-portforwarding-ranges\nRelated-Bug: #1885921\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1dde6c24b777a86eb1a562a4ee2ca5019ac1e08a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1dde6c24b777a86eb1a562a4ee2ca5019ac1e08a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
