)]}'
{"id":"openstack%2Fneutron-lib~832832","triplet_id":"openstack%2Fneutron-lib~master~Ifa72655b272bb466fff5fd9c84f6b833a673e86c","project":"openstack/neutron-lib","branch":"master","topic":"bug/1886798","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-03-11 17:06:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifa72655b272bb466fff5fd9c84f6b833a673e86c","subject":"Rehome ``NumaAffinityPoliciesEnumField``","status":"MERGED","created":"2022-03-09 17:18:16.000000000","updated":"2022-03-11 17:08:57.000000000","submitted":"2022-03-11 17:06:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832832-bug/1886798","meta_rev_id":"e48d26507d7c731f622f94b9d9a9afdc0c2320f3","_number":832832,"virtual_id_number":832832,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-11 17:06:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2022-03-11 15:13:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-03-10 06:55:50.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2022-03-11 15:13:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-09 19:02:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-09 20:06:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2022-03-10 06:55:50.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-03-11 15:13:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"7c8e6d2c44e4d2899e3029826f108a1986cd7216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-09 17:18:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cff44d87348fa68b4e38a5c4a380ce0d863615d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 19:02:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f4a0abb9bfe408081e301523c0ceab5 : SUCCESS in 6m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/448e5ac0f5724b75b950bbe2be0ca75f : SUCCESS in 6m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b4f36e365eb47dabfb78c4ccceb4383 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27b966a06b8f4b77a0729361e7573735 : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6911b1cf63cc4a7f84440fd1acb0a7e1 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a3dcc46a1594cb6af92f411a84f2f41 : SUCCESS in 7m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/814e0b211b1d4b1fb4e409c3d3919008 : SUCCESS in 1h 37m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeaf3a73685c4db5b164555deeb98e28 : SUCCESS in 5m 15s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2cbe4e696812400e82f998283f34dd32 : SUCCESS in 47m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3a4ee453d024e479cf8b9bbb5af167f : SUCCESS in 49m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4caf78c61720af4194acf11f7942d54424c483e7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-03-09 20:06:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3c142d21afc47c7cc62f7dc5da9ac740ee19e91","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-03-10 06:55:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4f55ab73d348f30f8f70b3bba73bfb0b8590af2e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-11 15:13:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ad8dc1a39ffb374fbf1cd9c013392e0b0fe3d79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 15:14:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d77b2227c96c467615372e4b9d3ca8fecc5176c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 17:06:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c38d605815cc456c9d464b93936d0664 : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11ad5c9f08744332bb1c018147d721dc : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9927cb5c3ef4421ebd5bbe736c58c11f : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf1bdf19c3934dabb45b196a503acee8 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5e9ed647d234905969a5e4a0b2b0cf6 : SUCCESS in 6m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4224e77f4b24a7a920dacdec2f98baf : SUCCESS in 1h 49m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0773ef10b0fd40b3b1da6fbaecb37ae1 : SUCCESS in 5m 12s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/30d45315bb9c40a9b108c3ef37ac6faa : SUCCESS in 54m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/453a79f45409443692872aafa9b21e9d : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ab76d017e53a68b542b4ec6d97451c653dd77c66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 17:06:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e48d26507d7c731f622f94b9d9a9afdc0c2320f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 17:08:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b516704bb7749f6ab9bceb4f7459ac1 : SUCCESS in 1m 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8ed3fdc5e3b4a828b8c7f2698216bc0 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"096fb92766007a6cdb01d94d26217c54c1039b30","revisions":{"096fb92766007a6cdb01d94d26217c54c1039b30":{"kind":"REWORK","_number":1,"created":"2022-03-09 17:18:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/32/832832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/32/832832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/32/832832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/32/832832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/32/832832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/32/832832/1"}}},"commit":{"parents":[{"commit":"823be18b60e2e17129bb94f6c176e370f9eca427","subject":"Merge \"Support binding floating ip to smart-nic port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/823be18b60e2e17129bb94f6c176e370f9eca427"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-04 22:28:54.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-04 22:31:39.000000000","tz":0},"subject":"Rehome ``NumaAffinityPoliciesEnumField``","message":"Rehome ``NumaAffinityPoliciesEnumField``\n\nChange-Id: Ifa72655b272bb466fff5fd9c84f6b833a673e86c\nRelated-Bug: #1886798\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/096fb92766007a6cdb01d94d26217c54c1039b30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/096fb92766007a6cdb01d94d26217c54c1039b30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
