)]}'
{"id":"openstack%2Fneutron-lib~833103","triplet_id":"openstack%2Fneutron-lib~master~I7f69d66d3d55c8cf44785c63d215c9f604b9ef52","project":"openstack/neutron-lib","branch":"master","topic":"bug/1964575","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-04-18 15:17:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7f69d66d3d55c8cf44785c63d215c9f604b9ef52","subject":"[WIP][DNM] Refactor session \"is_active\" handling for sqlalchemy-20","status":"ABANDONED","created":"2022-03-10 15:05:11.000000000","updated":"2022-04-18 15:17:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d435d4e81b551d82343983e9bdb632d009701fdb","_number":833103,"virtual_id_number":833103,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-11 13:19:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 17:23:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dbf15f99853f76a7fcb0024c62efbbb5306c4c3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-10 15:05:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d55f1e7ba3e972820286107f704c9c593ab5bfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 17:23:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ce39a5422a3403da738ada73e6bbecd : SUCCESS in 1h 56m 03s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8c2590279e704e2c8d2d8ff0be38d1aa : FAILURE in 16m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8ee94bb214493d46809e520284a11bafb90aedb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-11 09:28:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b8bcb6b619fcbf83a1852b14ba6489bb6bc7d43b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 11:12:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/acefd517c0574fb183b17cbedd02a31b : SUCCESS in 1h 40m 34s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/74542dd1b9d54e5b8f84bcf6e737f936 : SUCCESS in 54m 21s","accounts_in_message":[],"_revision_number":2},{"id":"27126c4e28a8b1da5cc5e5f86ff2e195b02a9e08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-11 11:33:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e4504bd3b151ab898f2151ba099b3d5819b8e59a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 13:19:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5542f8ff1f8f44c5b3e9849a6512b9ae : SUCCESS in 1h 31m 53s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/66fa973f16d3469ea2e7a2afcce485e2 : SUCCESS in 50m 54s","accounts_in_message":[],"_revision_number":3},{"id":"d435d4e81b551d82343983e9bdb632d009701fdb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-18 15:17:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7761ec6bd2d577c386240e4c6f5655f96a555011","revisions":{"19f3688d788238b8162bee8dc141357368a1a260":{"kind":"REWORK","_number":1,"created":"2022-03-10 15:05:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/03/833103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/03/833103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/833103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/833103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/833103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/03/833103/1"}}},"commit":{"parents":[{"commit":"6dfa4f5dd6b672d529f891e67da1377969d687cc","subject":"Merge \"Add filter flags for QoS rule types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6dfa4f5dd6b672d529f891e67da1377969d687cc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-06 09:54:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 08:00:07.000000000","tz":0},"subject":"[WIP][DNM] Refactor session \"is_active\" handling for sqlalchemy-20","message":"[WIP][DNM] Refactor session \"is_active\" handling for sqlalchemy-20\n\n---------------------------------------------------------------\n[2] AUTOCOMMIT\u003dTRUE AND WARNING WHEN EXECUTING TXN WITH AUTOBEGIN\n---------------------------------------------------------------\n\nSince sqlalchemy 1.4, \"session.autocommit\" is False by default; in\nsqlalchemy 2.0 this will be the only value accepted.\n\nThe ``_orm.Session`` is considered active when [1]:\n- there is a transaction and this transaction is active\n- there is no transaction [2], the class ``_orm.Session`` will\n  autobegin when it is first used.\n\nThe second one breaks the way Neutron considers a session is active:\nonly when a transaction is in place, Neutron considers a session is\nactive.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/828739\n\n[1]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3918-L3950\n[2]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3930-L3932\n\nChange-Id: I7f69d66d3d55c8cf44785c63d215c9f604b9ef52\nCloses-Bug: #1962153\nTopic: sqlalchemy-20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/19f3688d788238b8162bee8dc141357368a1a260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/19f3688d788238b8162bee8dc141357368a1a260"}]},"branch":"refs/heads/master"},"1f671b2eb3128aa2fef447c1eb278ae9271943d2":{"kind":"REWORK","_number":2,"created":"2022-03-11 09:28:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/03/833103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/03/833103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/833103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/833103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/833103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/03/833103/2"}}},"commit":{"parents":[{"commit":"6dfa4f5dd6b672d529f891e67da1377969d687cc","subject":"Merge \"Add filter flags for QoS rule types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6dfa4f5dd6b672d529f891e67da1377969d687cc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-06 09:54:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 13:46:37.000000000","tz":0},"subject":"[WIP][DNM] Refactor session \"is_active\" handling for sqlalchemy-20","message":"[WIP][DNM] Refactor session \"is_active\" handling for sqlalchemy-20\n\n---------------------------------------------------------------\n[2] AUTOCOMMIT\u003dTRUE AND WARNING WHEN EXECUTING TXN WITH AUTOBEGIN\n---------------------------------------------------------------\n\nSince sqlalchemy 1.4, \"session.autocommit\" is False by default; in\nsqlalchemy 2.0 this will be the only value accepted.\n\nThe ``_orm.Session`` is considered active when [1]:\n- there is a transaction and this transaction is active\n- there is no transaction [2], the class ``_orm.Session`` will\n  autobegin when it is first used.\n\nThe second one breaks the way Neutron considers a session is active:\nonly when a transaction is in place, Neutron considers a session is\nactive.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/828739\n\n[1]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3918-L3950\n[2]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3930-L3932\n\nChange-Id: I7f69d66d3d55c8cf44785c63d215c9f604b9ef52\nRelated-Bug: #1964575\nTopic: sqlalchemy-20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1f671b2eb3128aa2fef447c1eb278ae9271943d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1f671b2eb3128aa2fef447c1eb278ae9271943d2"}]},"branch":"refs/heads/master"},"7761ec6bd2d577c386240e4c6f5655f96a555011":{"kind":"REWORK","_number":3,"created":"2022-03-11 11:33:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/03/833103/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/03/833103/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/833103/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/833103/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/03/833103/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/03/833103/3"}}},"commit":{"parents":[{"commit":"6dfa4f5dd6b672d529f891e67da1377969d687cc","subject":"Merge \"Add filter flags for QoS rule types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6dfa4f5dd6b672d529f891e67da1377969d687cc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-06 09:54:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 16:02:45.000000000","tz":0},"subject":"[WIP][DNM] Refactor session \"is_active\" handling for sqlalchemy-20","message":"[WIP][DNM] Refactor session \"is_active\" handling for sqlalchemy-20\n\n---------------------------------------------------------------\n[2] AUTOCOMMIT\u003dTRUE AND WARNING WHEN EXECUTING TXN WITH AUTOBEGIN\n---------------------------------------------------------------\n\nSince sqlalchemy 1.4, \"session.autocommit\" is False by default; in\nsqlalchemy 2.0 this will be the only value accepted.\n\nThe ``_orm.Session`` is considered active when [1]:\n- there is a transaction and this transaction is active\n- there is no transaction [2], the class ``_orm.Session`` will\n  autobegin when it is first used.\n\nThe second one breaks the way Neutron considers a session is active:\nonly when a transaction is in place, Neutron considers a session is\nactive.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/828739\n\n[1]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3918-L3950\n[2]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3930-L3932\n\nChange-Id: I7f69d66d3d55c8cf44785c63d215c9f604b9ef52\nRelated-Bug: #1964575\nTopic: sqlalchemy-20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7761ec6bd2d577c386240e4c6f5655f96a555011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7761ec6bd2d577c386240e4c6f5655f96a555011"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
