)]}'
{"id":"openstack%2Fneutron-lib~833941","triplet_id":"openstack%2Fneutron-lib~master~Ide8229f101456bc448f4acc78ff74b879809261a","project":"openstack/neutron-lib","branch":"master","topic":"bug/1886798","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2022-03-22 14:09:06.000000000","reason":"Brian Haley replied on the change"}},"hashtags":[],"change_id":"Ide8229f101456bc448f4acc78ff74b879809261a","subject":"Add NumaAffinityPoliciesEnumField class unit tests","status":"MERGED","created":"2022-03-16 00:35:51.000000000","updated":"2022-03-22 17:24:56.000000000","submitted":"2022-03-22 17:23:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833941-bug/1886798","meta_rev_id":"6da351104c1ce6b4960fd23c4b6254071050df0e","_number":833941,"virtual_id_number":833941,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-22 17:23:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2022-03-22 11:04:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-03-17 07:04:08.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":2,"date":"2022-03-17 23:41:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2022-03-22 11:04:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-16 01:49:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-16 02:20:19.000000000","updated_by":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"CC"},{"updated":"2022-03-17 07:04:08.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-03-17 23:41:59.000000000","updated_by":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"REVIEWER"},{"updated":"2022-03-22 11:04:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"732022f6fb9baafb3c99d0eac5bdf6e8d989fad1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-03-16 00:35:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"320d600a016599d8245aee7e952376982234e9ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 01:49:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb16c3c1fe7940c89706d718032ac19d : SUCCESS in 7m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fccc68030bd44d49cdc48a571a0988f : SUCCESS in 6m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa30c17c30ca42d7af2f29b674109d7e : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29020f5c80a4b5d8f5acc917f4a1909 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc8484ac3f1a431d8c7aa0c43a454120 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8e3c87096b146c9913961f8ce46e04d : SUCCESS in 5m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7aca55d9eef24640ae6db2aeb9d018bf : FAILURE in 1h 12m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/23d11bb2f5964c05b582a20a9b8aba08 : FAILURE in 45m 48s","accounts_in_message":[],"_revision_number":1},{"id":"99a3afffc9476aa36f535d39c6435192b9210f7e","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-16 02:20:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"345ebbaf23ba9f7bc40934e04e6bc09a2679cbfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 04:00:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d11698a82c4bbbbbff0daf7c2d5507 : SUCCESS in 7m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efb180a69002450d860eaab0c48dc482 : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94b9ae1396824354bc082e6ddf163ac4 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/480ba794d3a94bc3b55ae86c667c72b0 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c09d098fbb2846de97519d6f22d6f7ff : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45ac6c75edcf4b2da495518f3fcb9a26 : SUCCESS in 6m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a6e002f68a7486a88364ee0f1d855ba : POST_FAILURE in 1h 31m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1366ed489282474db85bd2d27bc5d79d : FAILURE in 56m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1340345489904e72e810c870ef632af3892dc906","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-03-16 17:52:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f72613a36e7889d0c7a1ff6399f1b050520c88fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 18:59:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f279af3eb33948759860040915a72610 : SUCCESS in 7m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ee4725ab1af47be8aaec4abf1a79f3a : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cd790b573684873a3800160e368abe9 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/493d04ce2c9b4048869a278683fc2b2e : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bea4315ce36c428ca30eef3727a8e4ed : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/998f879e9859447fabeb5c0446a233be : SUCCESS in 5m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b235b89ca5a4f10b584c905e937534f : SUCCESS in 1h 02m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e6bdd5d236c4429e84ae6a76ae07b1e5 : FAILURE in 41m 57s","accounts_in_message":[],"_revision_number":1},{"id":"1ba9a12a4e60f0386c394ef8f6a22b69bb6dad64","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-03-16 19:59:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6fedfc749a3436ecdcb2f67d71ddb2e60352e44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 21:25:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48110165b96e4242a09a68a5a1250359 : SUCCESS in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a1cc412957e417c87c45679ec114466 : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24066aab1e2348bca1dfd94a957fe2c3 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa83f978379040a9b2e2a93a7da4665b : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15ec8030ccab40ef80e14f7ac27b3c8f : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57d15ff9967548dbb6edcc01a87595c7 : SUCCESS in 5m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52b7125062d34383b9160e77785aa710 : SUCCESS in 1h 20m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8f848a6d7b245209e791711730b8c57 : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":1},{"id":"40b6f03e45e9bd303334af1c961ed6c2d02d3dc2","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-03-17 07:04:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ba685dd1c506b91dd046638ad529a890d6b94813","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-17 23:41:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3d428fb5ffba282477a0f9159cd346fadabc8962","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-22 11:04:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4dbcf6c3faba9b0b46764e8c429370a69143623b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 11:08:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b51ca1d3b07ed894c3c69bd3ec39574b1820d81b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:48:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8699a34803704c4a8caae94e5331b8bb : SUCCESS in 7m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b6e34e2430f4b4285c487d1bacdd3d4 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f96bab9b3ff4b36ad506420c415bda4 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51849573690f4c70beebdef8c7b3b7c1 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bacca7dbde6417e8937a7db75ea00b9 : SUCCESS in 5m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c87bc107ee9b4eb184afcf8e2878362d : SUCCESS in 1h 33m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/63e7b3b7e8094a16ac4702af087eefac : FAILURE in 53m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b3e1c06b1e7a38aa61d9f0c8c866e98b9fea4a84","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-03-22 14:09:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7192bd2d7b923d53a98e45f78a6fca0a12ef7449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 15:31:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9019c54059a14dc28ecac92abd354142 : SUCCESS in 8m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/736058c5b47948eca8aebd570b05a365 : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da78428971f0430f805f548a9a655851 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5afa196387c147c2bc758f90ca522a2d : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4285d7bab2fd48f88bda92d151b74135 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd79317e8195406e8ba66454ee868f32 : SUCCESS in 6m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/002cbce95658416caf9e8ddb35736d9e : SUCCESS in 1h 19m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a04bff06b7aa4a0e8b7d251839ad1c12 : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ed4fb886470791852994c42a6574115d6d7c3f24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 15:32:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d2a9f00cd2681ef8d4b785366edba9b9e0e1031","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 17:23:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9a6e2026a4248a3b2dfe953854edc15 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07dfa21551424cc6a8f37c67a89f676c : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87932fb6289d45018e4b2feeec9ee22a : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f392994570584a51986044c3b9452db2 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c04ae1b5679740ccb0ed0606deaeea44 : SUCCESS in 5m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f189bb976dfd429bad2ffc88aa185821 : SUCCESS in 1h 50m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d824bb3e38648a0861f52e2bcd02afa : SUCCESS in 51m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5edd84fef164f663958a27cffaa057646012579f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 17:23:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6da351104c1ce6b4960fd23c4b6254071050df0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 17:24:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39b7098b23764ccb89e7afbf2d11a0f0 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ef0d26ad372031c4118304f29bebb963a7981f4","revisions":{"5ef0d26ad372031c4118304f29bebb963a7981f4":{"kind":"REWORK","_number":1,"created":"2022-03-16 00:35:51.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/41/833941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/833941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/833941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/833941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/833941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/833941/1"}}},"commit":{"parents":[{"commit":"096fb92766007a6cdb01d94d26217c54c1039b30","subject":"Rehome ``NumaAffinityPoliciesEnumField``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/096fb92766007a6cdb01d94d26217c54c1039b30"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-03-16 00:34:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-03-16 00:34:25.000000000","tz":-240},"subject":"Add NumaAffinityPoliciesEnumField class unit tests","message":"Add NumaAffinityPoliciesEnumField class unit tests\n\nPrior change rehomed this class here, add some basic\nunit tests to cover it.\n\nChange-Id: Ide8229f101456bc448f4acc78ff74b879809261a\nRelated-Bug: #1886798\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5ef0d26ad372031c4118304f29bebb963a7981f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5ef0d26ad372031c4118304f29bebb963a7981f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
