)]}'
{"id":"openstack%2Fneutron-lib~835127","triplet_id":"openstack%2Fneutron-lib~master~Ia261d717645606e458b826664037895fa07e646a","project":"openstack/neutron-lib","branch":"master","topic":"trivial-log","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2022-03-28 15:51:19.000000000","reason":"Brian Haley replied on the change"}},"hashtags":[],"change_id":"Ia261d717645606e458b826664037895fa07e646a","subject":"Remove unused code","status":"MERGED","created":"2022-03-24 21:41:20.000000000","updated":"2022-03-28 18:53:47.000000000","submitted":"2022-03-28 18:52:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835127-trivial-log","meta_rev_id":"a4d08209d37e0a02b22128b4602ae8103618fdbe","_number":835127,"virtual_id_number":835127,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-28 18:52:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2022-03-25 15:42:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-03-25 08:19:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":2,"date":"2022-03-25 05:38:07.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2022-03-25 15:42:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-24 23:10:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-25 05:38:07.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-03-25 08:19:04.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2022-03-25 15:42:39.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"246e2ffd6ebaa05c836577a76546288d40ce8422","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-03-24 21:41:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"630951f55be07eb917702626fbec28210bb4648c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 23:10:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db85286f085416aaafb3f93a661c5cd : SUCCESS in 8m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0c547d4694e439392af54ec741e92b0 : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d29dbcdde9a40129b7211a2505bc878 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/026730242b1542aab8624432bfc9be34 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f683328f901403992ba4c6fd122c093 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffc4fcb475a14e78b652e285c87be8af : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/beb7c57121f44beb97855f879eb628b3 : SUCCESS in 7m 38s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2a7e923ca414af4a9aba8d99c03f275 : SUCCESS in 1h 27m 00s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c9ac870ae69641ecbb612a7e20c63bee : SUCCESS in 45m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5cb5091fcd842b893a4afa3ed7be0d7 : POST_FAILURE in 53m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5333162cac84b2a8b70c7146468cda56dafc3f1a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-03-25 05:38:07.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b147e7734a1c5c2e045a2969002ce877085fdcae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 07:09:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e839ff595fc445d295f7194293925757 : SUCCESS in 7m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81b48a2464434f8fb9980003c4e3e578 : SUCCESS in 6m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0c2a9d2afe64304b9c34238a895c85d : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af2f5db18bf145bf9dbbc6250329a14d : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b19e440bf5f446da7244c63aebecb01 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1772da72ccf74e738b7b1bbedd14a022 : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7782752047ce465886ca1a6bc16f469e : SUCCESS in 7m 48s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1862e7d3bc874330b3964614676e0273 : SUCCESS in 1h 29m 51s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b358a96ac0734a67baf712caafc2ffbf : SUCCESS in 45m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/35786849ab9247fb9f34ac3aa5bfa116 : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":1},{"id":"71cd6bc462960ef541aa828f095af675300a634e","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-03-25 08:19:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b1f011487eb330754944c0724efc901ba2d1e8d7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-25 15:42:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"efafa789fbe9568d43b0c0ba2ac4ace012d686a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 15:43:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91aa104aa995f41fd1220481d08e16a09e1b4aad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 16:29:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbceaf91c2cb42719f1a5111652789ed : SUCCESS in 8m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bc32d0f33f74c9684c394b463bccf71 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e22cb7813bd44ccdbf0fe618f3215be3 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/334da28ada7c45c485be5577c243b2cd : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6af3c5aaed62455598eb2780b6e6a823 : SUCCESS in 5m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fdbfe30146f4a9c9a4e847817e8761c : RETRY_LIMIT in 16m 15s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fa55e1aea2b146dd9b2caa3f48d9f0ee : RETRY_LIMIT in 11m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a0540487785c44eeb80580bd3050fd51 : RETRY_LIMIT in 16m 35s","accounts_in_message":[],"_revision_number":1},{"id":"98b43a5d0a983b744442e1f5f2f2206410938b6e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-03-25 16:55:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aace2ebadd9884374a343d763e5b3708de7028f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 17:46:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4daab01c31514c51920d3dc14eef3f56 : SUCCESS in 7m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1de30a2f0b845fcb9cc216e80d17763 : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29bb1fc48fb5406781367904d2397ce1 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb8f0ceb0f274808a263e40a7d3ddfef : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/331e443b6fa14089a4a9025194986e27 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab81c0c1931b46d0adb9436dc347f533 : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c607f60723bd4526bcfb7552201ba42d : SUCCESS in 8m 28s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89f93649a3454a35aa543aa8492bb7d8 : RETRY_LIMIT in 8m 31s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4ef80f045bdb4b089e1acb7333f27818 : RETRY_LIMIT in 8m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/14a8b3793c8f4b25ae6349ada5ec964b : RETRY_LIMIT in 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6165e5172d80357593207cdc67f6b64ae224937e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-03-25 17:50:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b668ba7f84d52c10587af6e5ce8514af8eb9a26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 18:34:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f22d3713625408ea0faafb8dddba910 : SUCCESS in 8m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31dfc12ed7314d5c8fab8125db783063 : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fe174797684435b8db25ac4ceed49ab : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dddc2177dbf34c87a9bf1cf6d0290743 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2401033e4f54441a8bf705446117017 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73b312bf825f4e58ac75c47b7a111b5b : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d01045a751e940639971b298c92a606f : SUCCESS in 9m 17s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44b60ded5eb54aecabfd43e7907d43e0 : RETRY_LIMIT in 8m 01s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4f46018517724a0b869990059fe26768 : RETRY_LIMIT in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ac724b945ef482095ed20373892c4ee : RETRY_LIMIT in 6m 57s","accounts_in_message":[],"_revision_number":1},{"id":"398791d3e91139c4a5494113da91d1e023eccec4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-03-25 20:40:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c32a42757f2e88d891210404ed472d3eaadcd1b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 22:05:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6d3fceebb814088956be984dc47f3ee : SUCCESS in 7m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfdebe39dc4f46cf97e275d792a343d7 : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a63099980b9c4a8f99083ad4df675134 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ad2107f9d164ded97e9b748bcc52fb4 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f02c172b8aca4bc5a9526b6f2a5dbfb8 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4abebf17cd5c4c568eee41fb8a6e7221 : SUCCESS in 8m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02fcafc86866404799f0766d733afec0 : SUCCESS in 8m 48s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08f9bab3a62147838d699a22bb4fa63f : SUCCESS in 1h 23m 39s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f4e19e951e6845d9858c5f315e2a2e68 : SUCCESS in 40m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/853ace9814354585aaad3709818fb356 : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3da3eab54dd8897aba3a538d63e65c6ca63b09a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 22:05:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4aaef57e2d122a150346d05c04250e5b8fa74256","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 23:27:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6dd88b569a44a70adaacc4d35b3cfa5 : SUCCESS in 7m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecbfe6de1f1a4ae4a026b967dd87e773 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55abfb9cb5a142789fb58dd3295a3c06 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6185b748fd8e46fa9302ef721f7d60ac : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0f1c43d7643433aaf8d46b8b8d1a217 : SUCCESS in 5m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22eb1a9102db4e33b447cee677550b67 : SUCCESS in 1h 19m 54s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/93525e19d70e4e96b0d47183346829f9 : SUCCESS in 50m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e7379419ed14449ac394982ee5d092b : FAILURE in 51m 23s","accounts_in_message":[],"_revision_number":1},{"id":"603b233e60445ca137635a1cf49da995009d422f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-03-28 15:51:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d862eb6f4fc4e690bb42316b3691b66ad726dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 17:27:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca00f425458345d6afa815e9f28eb98f : SUCCESS in 7m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdad87a1005546af8f2a6e6b285e4670 : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3692b4b737ad46bdac33069ee859c7c7 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a285c517246441929989807e92d3495c : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35516f0c538d4f5ba6b802d87794cef3 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63d1831f61334aabbef74031303d3fdd : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0be5447e5154081a54e422fabd80971 : SUCCESS in 8m 49s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49d76d56fb8a4e8fb9696de83b67a9ce : SUCCESS in 1h 29m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/80981de0a16845719cfd8d7112364e2b : SUCCESS in 53m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f14a5ca3ae0c4b3c915ef9703a179957 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b66b44985c761a99df1039ccf380187c096f2991","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 17:29:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e88c804e549225cff1e091c862bae78f246f431","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 18:52:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60d69727ddb4fa3932c3219861d3d16 : SUCCESS in 7m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d5d969ff93442638a55287d40513684 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0385ad1ac20b4dbaa82a2e1c7310233c : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9546d40f2fa4647b3b9e8848e479bbf : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57d810cceb44436792182f56a3fba0d7 : SUCCESS in 6m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d1e02fe611245bbb84b2c3eff2f8f53 : SUCCESS in 1h 21m 52s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e98f16530a4c47dfa37ab4eaf6facd12 : SUCCESS in 44m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9071564efa48462abb074f18877d25f4 : SUCCESS in 51m 47s","accounts_in_message":[],"_revision_number":1},{"id":"98169999e4b732e16655280005e1598477e5fa18","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 18:52:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a4d08209d37e0a02b22128b4602ae8103618fdbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 18:53:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/967bc1925f184f63aa61227b4a8b876f : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3ee9e06c2492abce53d4bb8d04bb3fdbbe5a5ac","revisions":{"a3ee9e06c2492abce53d4bb8d04bb3fdbbe5a5ac":{"kind":"REWORK","_number":1,"created":"2022-03-24 21:41:20.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/27/835127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/27/835127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/835127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/835127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/835127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/27/835127/1"}}},"commit":{"parents":[{"commit":"74d3765407de58fcf871f7de48aaed544feedabf","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/74d3765407de58fcf871f7de48aaed544feedabf"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-03-24 21:39:44.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-03-24 21:39:44.000000000","tz":-240},"subject":"Remove unused code","message":"Remove unused code\n\nThere were a couple of files that import from oslo_log\nbut never actually used it, remove them.\n\nTrivial-fix\n\nChange-Id: Ia261d717645606e458b826664037895fa07e646a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a3ee9e06c2492abce53d4bb8d04bb3fdbbe5a5ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a3ee9e06c2492abce53d4bb8d04bb3fdbbe5a5ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
