)]}'
{"id":"openstack%2Fneutron-lib~835234","triplet_id":"openstack%2Fneutron-lib~master~Iad1e2c9f797091d728d419c6b9dc67d861d4214a","project":"openstack/neutron-lib","branch":"master","topic":"bug/1959333","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-04-11 10:50:19.000000000","reason":"Slawek Kaplonski replied on the change"}},"hashtags":[],"change_id":"Iad1e2c9f797091d728d419c6b9dc67d861d4214a","subject":"Add oslo_policy.InvalidScope exception to the api faults map","status":"MERGED","created":"2022-03-25 13:52:52.000000000","updated":"2022-04-11 13:45:33.000000000","submitted":"2022-04-11 13:43:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835234-bug/1959333","meta_rev_id":"ce29f5fd041aa806c58db66e77fd51e8ed4ab849","_number":835234,"virtual_id_number":835234,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 13:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-04-04 13:47:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-04-01 09:56:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-03-28 06:57:53.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-04-04 13:47:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-25 14:48:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-28 06:57:53.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-04-01 09:56:30.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-04-04 13:47:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"520cec631cc79f13ae3695720fc2f2dc47e940d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-25 13:52:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"417ebb8e9526c4bb18639e0c3e30b99a402e73b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 14:48:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb8a9601028548fd93fe550e1c455b5d : SUCCESS in 6m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc6b08f989bd45a996db82d58b4b2ff3 : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a8b924f0e6d43c4a16ef89b5245118a : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12e43865bc324e2c80c6638ff5cc3143 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/413ebd893c644d0b9fda4f9ce41bd4f7 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f730953d344405ca3d4a8a3bf06f041 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0870bb638883488db98eb8dd8aeef17e : SUCCESS in 7m 38s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8251a0c6f579477e8138c23c6de5a20f : RETRY_LIMIT in 11m 12s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b92b474a0dce406db429b27e3d459841 : RETRY_LIMIT in 15m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e89b77b3491448cbb3db34253e7d3e0 : RETRY_LIMIT in 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a2922152904d32a3dc5b1089f5cfeeefc98fa724","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-25 15:26:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1f0b0686c8cc016f6f83946dca5b9fa911b44c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 16:37:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88420f13872c4bb4b0e6d5c3546787af : SUCCESS in 7m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/262d598166d341628164285b1a92fa7b : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adba0da7cbec439e8173d96866a779a5 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98f6b5e86ddb4a53b7a9b9a21cc4411c : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d32c60716d30472dbd24956f9541778c : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0322fa81e6b9481f9c5d25ce17766434 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8faab3c58b942e6b43f670a8b116fd1 : SUCCESS in 8m 15s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34e42daad42845aa90d3bf0e37444974 : RETRY_LIMIT in 17m 05s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/59179fc8e3e84ce3927544d71f417e70 : RETRY_LIMIT in 8m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a6e277c42b44b30acd315d4949940a0 : RETRY_LIMIT in 8m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1f867cc2160653f5dc774f09ba08ddc450e16cd4","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-03-28 06:57:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"31526eea308855493af68a44d6b506cc21d0cd85","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-04-01 09:56:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7005db68acd57d1f775ad02fdbe5b0d23c099bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 11:46:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f4a601700514670be7b7831a31c44bd : SUCCESS in 6m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51835f39d531492f9bde55e52962326c : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/630f12b3ee964fdea7c7fb8b62db8e50 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bd6fe2a8108456c873a74b10ab09763 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d588e51d1b764bae80226c8e9e7cf126 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de2ab964c4e54dd090b7bd705648da4c : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c6fd050f62246ad957c226e4da71fa9 : SUCCESS in 6m 38s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b10928bbe0c4ab29e03ac35f5137f7a : SUCCESS in 1h 42m 55s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f3f64dde89bc427a8301f197b2977163 : SUCCESS in 51m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf29873a45c0481ba5837475a96fa740 : SUCCESS in 55m 24s","accounts_in_message":[],"_revision_number":1},{"id":"1559ff772e338ca27c5a9a17d45bffc10734db5e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-04 13:47:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d46219fda31cef37518e72f6a843c537f8e8b1fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 13:48:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b0a328bdc2d4649a5c6f244b1b0409622ef3282","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 15:17:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8ba98991c9c4a94a7243f37530f4e05 : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/defef95226a6473b9a176b9b5e8737a8 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf253bf50b44bb5861762cb2c66c83b : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d43497106d4c4b39805d23f4fd1ff12f : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4264f1b7d3a744e6802492f7c2f8d0a3 : SUCCESS in 4m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89945e0ce966405b86e5f88865bd22b6 : SUCCESS in 1h 28m 17s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/af4135dd2ed942668ce49045f4dd3368 : SUCCESS in 38m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef6e43abd900483891e0a632c7518ce9 : FAILURE in 52m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a30ab6e3b42b30c6603f9bcf6a9684623aee60fa","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-05 15:10:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e30fa854930bc0fcd9a74af95563788af42a6b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 16:06:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96f49120ebe0443d9617cbd969057718 : SUCCESS in 6m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c3f6e2e376848198e5074a4c99b1b03 : FAILURE in 3m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65093212f6944e11b42af91087bed844 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a0d939b2eaf4f898dfb7ce738f45a03 : FAILURE in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c52ed33cd84b4b3a8ff50115a332cb95 : FAILURE in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68df924850ea41f6ac9214593c22fdd3 : FAILURE in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3e9b1674dbd401d9aa679a49682684b : FAILURE in 4m 34s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5600c3b95f2046d5839ea3ac74956047 : FAILURE in 11m 12s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f7ff2e101a674b90a4d9583d95d77c72 : FAILURE in 17m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bacccf8fae444ee3af77337dab8e9f48 : FAILURE in 54m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ca981f91a82fb633615f04389e1b153cdb771829","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-11 10:50:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a21a38f3629839e28f0d1e6467e78d95437d115","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:08:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f07d54e29540f78bf4f3e9bb4db4db : SUCCESS in 5m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05988355b8454fd88efbae5fdbafd58b : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73cba229d41346d9ba77873d267529f6 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4472a60e65ab4043b3f899c77779e71a : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd54b8148a6e43438aabe58c2b6d50bd : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c657581099b146328ee9892acb1f801e : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/063ea11ad929465eb188cc1fc1b961a1 : SUCCESS in 8m 08s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1319ce05576a40e487220c877aced086 : SUCCESS in 1h 15m 32s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/81b2158d12404776a5ce248e5833d279 : SUCCESS in 36m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/037e29ffd0fa4fbdae857d8b01ba20b3 : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6cb04cbabf44437323d0a6057a3e6f12cfb6611e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:09:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc9e5323348b52df88d3ffff64bad7f7746de499","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 13:43:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d24dfa4ff65490294ecee8b8dcd206e : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b4bcb3c3c924145b5df1041df753f42 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdad8133bfce47f3b0a0f888422d43ee : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21493897e37f4af994e87d846f8b1b20 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d861379be4694680904f6f26c55a8284 : SUCCESS in 5m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6465fd0a1aaf441696af6242924ec450 : SUCCESS in 1h 31m 50s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/aaf9d9805b604449914cd0ef0aa96c8c : SUCCESS in 40m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6039733cf35f454080b7ed1c02e08417 : SUCCESS in 1h 08m 15s","accounts_in_message":[],"_revision_number":1},{"id":"35572e0968d0c1797e09e3f79a2c952463af786f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 13:43:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ce29f5fd041aa806c58db66e77fd51e8ed4ab849","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 13:45:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df8011b6d3c48a2aae11aef7c9b1759 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da2baf389c85548b5c0b475a651aaf22bd7d9701","revisions":{"da2baf389c85548b5c0b475a651aaf22bd7d9701":{"kind":"REWORK","_number":1,"created":"2022-03-25 13:52:52.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/34/835234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/34/835234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/34/835234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/34/835234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/34/835234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/34/835234/1"}}},"commit":{"parents":[{"commit":"74d3765407de58fcf871f7de48aaed544feedabf","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/74d3765407de58fcf871f7de48aaed544feedabf"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-25 13:50:33.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-25 13:50:33.000000000","tz":60},"subject":"Add oslo_policy.InvalidScope exception to the api faults map","message":"Add oslo_policy.InvalidScope exception to the api faults map\n\nWith enforcing scopes enabled in Neutron, oslo_policy can raise\nInvalidScope exception while enforcing policy rules. So this exception\ntype should be handled in the same way as it is with\nPolicyNotAuthorized. Otherwise neutron returns 500 if InvalidScope\nexception was raised by the policy enforce.\n\nCloses-Bug: #1959333\nChange-Id: Iad1e2c9f797091d728d419c6b9dc67d861d4214a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/da2baf389c85548b5c0b475a651aaf22bd7d9701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/da2baf389c85548b5c0b475a651aaf22bd7d9701"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
