)]}'
{"id":"openstack%2Fneutron-lib~839956","triplet_id":"openstack%2Fneutron-lib~master~Ib5d4134bf59ea37d983f33807b396b31ffd9719d","project":"openstack/neutron-lib","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-05-30 11:25:21.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-05-30 09:46:59.000000000","reason":"Lajos Katona replied on the change"}},"hashtags":[],"change_id":"Ib5d4134bf59ea37d983f33807b396b31ffd9719d","subject":"Check proper config option to see if scope is enforced or not","status":"MERGED","created":"2022-04-29 15:02:50.000000000","updated":"2022-05-30 11:27:14.000000000","submitted":"2022-05-30 11:25:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839956-secure-rbac","meta_rev_id":"4dfcad62447ea5c7fb7fca5ac9c525a6d85d5cdc","_number":839956,"virtual_id_number":839956,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-30 11:25:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-05-30 09:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-05-30 09:46:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-05-30 09:53:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 16:47:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-29 18:15:20.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2022-05-02 09:21:41.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-05-30 09:53:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"53e54ee850e0094805b17c994fd7b29c9f8e5b70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-29 15:02:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05fd29e92f1c0366d2844f0a224b902976e6f13d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:47:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ee25962ce0f4875b8283bf6edc57c94 : SUCCESS in 8m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe329f2a4fc44e23860302e414879a48 : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d91cabca5928418f97b3298f80a743d1 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd4f40f3aeda40828abe690e71ae7bb9 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c1dfe53224c417b90deff297d285aa4 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/375185cde2eb47089124059843454ded : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d62c8fe65924a0b8a72703cd4e3f11e : SUCCESS in 6m 59s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a3c208bac79415092e67f22dda41e06 : SUCCESS in 1h 37m 07s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/805a9ddebc03476384968a949e1d3c09 : SUCCESS in 52m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5708a7c373ad4d209ca7c6da03314dbe : SUCCESS in 1h 01m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/30b8c478e0304d8f9935f2f139df35f8 : SUCCESS in 56m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f747ac5f77a743bf97075a562423721f : SUCCESS in 52m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/38895adfab864af594dec43bbeed60be : SUCCESS in 52m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f05de2fc52404ecc95ac748527779af2 : SUCCESS in 30m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a95f4df712c74eb78342de67698f9eee : FAILURE in 59m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e8c6403c53d3f1deec9d616708ac3825a2b1a6ad","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-04-29 18:15:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec95ff198bbaddf1d0f6ce18d785aadae7797654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 20:28:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a41bea185504c19a1f6ec27909f279a : SUCCESS in 6m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37fbf46b90ac42898132d3c12576343c : SUCCESS in 4m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3bc4c40f7104141a1a94c6f636add98 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21ec766f8b924302b237ae41f8b0a4ed : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13bcd36293e8432abb3787d479cea313 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/689dfa8d41ee434a82cac9099ac9080e : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d79506080d7942928517154b177a0340 : SUCCESS in 6m 11s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab527aba119242a1853a954a41536f4d : SUCCESS in 2h 09m 08s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/73797fa504494e12a329f577ac5d674a : SUCCESS in 1h 18m 13s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/11421e70eb2c4b9a9e45f2d9901cd788 : SUCCESS in 1h 11m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a22829a977944cd8914a0e4346c2b310 : SUCCESS in 1h 20m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b5f34baf73c44a16952a7cb7d8c9dcd8 : SUCCESS in 1h 01m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3efacc55e7904272a07bf3d167536298 : SUCCESS in 1h 47m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/73e13e891b024a23bdd93c6c2df97357 : SUCCESS in 32m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f891bfa769484442861d5047aec7abca : SUCCESS in 48m 34s","accounts_in_message":[],"_revision_number":1},{"id":"92900125b96de193cdd6080ad27832d8e8080a68","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-02 09:21:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7167367dea136e9ff1399d2960bf665ea1613fc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-05-27 14:23:47.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a36a63823a61c6a843503e7dadece5d8273f3d52","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-05-27 14:24:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e98810baa7d68d41fcbc30ada596ebdc4a455e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 15:55:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/878a84caef074a308a7334430d3376f9 : SUCCESS in 6m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6927ed3e99a8455f92d8989b4c5c9a31 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c56eeccb407b4264972b0b82f271565e : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecd0f9f98712428cbf82df127b004c47 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e44939155a7a4a50b022ba1b26698d18 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d50d8ba0ab14ccca6095e534308cf38 : SUCCESS in 4m 42s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf0c3ea5f0b44aa58c41341f377794c3 : SUCCESS in 1h 23m 07s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c58593a00728466298cbb5f0cbc483f4 : SUCCESS in 1h 22m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0ce49f5a26e44073ba6b5037f1f399b8 : SUCCESS in 1h 14m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/44f64fc3169f4df5b337cb6abfd2ef60 : SUCCESS in 1h 09m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7178b9b189714b778b319b68a7038dde : SUCCESS in 1h 15m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e275d467e7bc41b59e717b6f154fbf2a : SUCCESS in 32m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ecfbfe3dc1064a73b92fa6d0f697eabe : SUCCESS in 54m 22s","accounts_in_message":[],"_revision_number":2},{"id":"3dd22982283a782674e5f704146e26b17fcaad80","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-30 09:46:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bcc89d232e0dd681d2b7dd952f30634e3be76e98","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-30 09:53:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c1dc6508b8c51d0ba30304b1769c8d23fd0934e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 09:53:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0be94f8928121a621407310e2f85609536404421","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 11:25:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61764564ef00436b8d8a55283e40b17d : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bac0cbc7f16544c0b1d7fa1542b0d163 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7d8655181c34a6c88b1da81f12fc118 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/609b54b6a601454ca5497b8b09a99dfc : SUCCESS in 4m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ab728a46dfc4173a7ddfe7c3e7fd3b3 : SUCCESS in 1h 28m 17s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/7424a690ae6e424688a2431e3c8166f8 : SUCCESS in 1h 29m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9f1a879344854596b300a23badb053d6 : SUCCESS in 1h 19m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8d0147fc52f6437d83a769bf403bb36a : SUCCESS in 1h 30m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/160f81cf6e9b4e78bb5e5ea6f211bf00 : SUCCESS in 1h 07m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/66dc6aad6b6d4a2fa0e3afb336f13fc4 : SUCCESS in 51m 37s","accounts_in_message":[],"_revision_number":2},{"id":"d36eb1692ac5b854812b46e3df812662b621d5d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 11:25:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4dfcad62447ea5c7fb7fca5ac9c525a6d85d5cdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 11:27:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0097349823b94cdfa128b7e929ba8c6c : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1de7dc1bdf914b1e34b4e604cbc810ef3181e286","revisions":{"473c71ba3f0d02e810c9bc5f318e7f3461eb61a0":{"kind":"REWORK","_number":1,"created":"2022-04-29 15:02:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/56/839956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/56/839956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/839956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/839956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/839956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/56/839956/1"}}},"commit":{"parents":[{"commit":"ccb996fbe5ebb542be954e13475271e2825e3615","subject":"Merge \"Remove unimplemented api-def: BFD monitors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ccb996fbe5ebb542be954e13475271e2825e3615"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-29 14:59:22.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-29 14:59:22.000000000","tz":120},"subject":"Check proper config option to see if scope is enforced or not","message":"Check proper config option to see if scope is enforced or not\n\nOslo policy introduced 2 config options which are related to the new\n\"Secure RBAC\" policies and are in most cases used together:\n* enforce_new_defaults\n* enforce_scope\n\nWhen I initially worked on new default policies in Neutron I always\nset both of them to True together and I wrongly added checking\nif enforce_new_defaults is set to True to check or set system_scope\nparameter in context object.\nAs system_scope is related to scope actually, it should be set/check\nwhen enforce_scope is enabled, not when enforce_new_defaults is enabled.\n\nThis patch fixes that by checking proper config option.\n\nChange-Id: Ib5d4134bf59ea37d983f33807b396b31ffd9719d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/473c71ba3f0d02e810c9bc5f318e7f3461eb61a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/473c71ba3f0d02e810c9bc5f318e7f3461eb61a0"}]},"branch":"refs/heads/master"},"1de7dc1bdf914b1e34b4e604cbc810ef3181e286":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-27 14:23:47.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/56/839956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/56/839956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/839956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/839956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/839956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/56/839956/2"}}},"commit":{"parents":[{"commit":"ccb996fbe5ebb542be954e13475271e2825e3615","subject":"Merge \"Remove unimplemented api-def: BFD monitors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ccb996fbe5ebb542be954e13475271e2825e3615"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-29 14:59:22.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-05-27 14:23:46.000000000","tz":0},"subject":"Check proper config option to see if scope is enforced or not","message":"Check proper config option to see if scope is enforced or not\n\nOslo policy introduced 2 config options which are related to the new\n\"Secure RBAC\" policies and are in most cases used together:\n* enforce_new_defaults\n* enforce_scope\n\nWhen I initially worked on new default policies in Neutron I always\nset both of them to True together and I wrongly added checking\nif enforce_new_defaults is set to True to check or set system_scope\nparameter in context object.\nAs enforce_scope is related to scope actually, it should be set/check\nwhen enforce_scope is enabled, not when enforce_new_defaults is enabled.\n\nThis patch fixes that by checking proper config option.\n\nChange-Id: Ib5d4134bf59ea37d983f33807b396b31ffd9719d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1de7dc1bdf914b1e34b4e604cbc810ef3181e286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1de7dc1bdf914b1e34b4e604cbc810ef3181e286"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
