)]}'
{"id":"openstack%2Fneutron-lib~840099","triplet_id":"openstack%2Fneutron-lib~master~I67453b5a0e6d41d5cb4ef0cbb47f955a448005eb","project":"openstack/neutron-lib","branch":"master","topic":"drop-l-c","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-05-02 13:27:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I67453b5a0e6d41d5cb4ef0cbb47f955a448005eb","subject":"Drop lower-constraints.txt and its testing","status":"MERGED","created":"2022-05-01 01:29:48.000000000","updated":"2022-05-02 13:29:48.000000000","submitted":"2022-05-02 13:27:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840099-drop-l-c","meta_rev_id":"86d976509b5aafe2df09549ea2324ba3c1f02e8e","_number":840099,"virtual_id_number":840099,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-02 13:27:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2022-05-02 11:28:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-05-02 08:53:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2022-05-02 11:28:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-01 03:13:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-02 08:53:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-05-02 11:28:34.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"2f0be5fdfb60c31b6ac01aac8ceeb21c914b6b3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-01 01:29:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79eab6b7ec4998bc11e93d911137351a3e75864d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 03:13:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/da90710f4cb54f2db6741072d45263bb : SUCCESS in 4m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9b93086c9ebf46fd8c3dd0df36b48de2 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfd6760bcabf4614a22a558ae2ac6887 : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/799b864bce374803a517ab69d6fa4b27 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d9b6947bb5d4b2da522b52959671f8e : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11cd3ed1f6cc45a3b32bca19a2c6405a : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92d50d4fc95246d49663a6aba265a3a6 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea63de43f7b44016b836da302fb1c924 : SUCCESS in 6m 32s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/686c91ec357f45118616f41bae830519 : SUCCESS in 1h 37m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26ba73884fc648dc8ab1527edbc38416 : SUCCESS in 3m 12s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a30e39c9c924437baee3fa8af3c00e77 : SUCCESS in 49m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4fef3d9e4f4f4d79aedc7d41e78c40d7 : SUCCESS in 1h 02m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b35acdb7f8564bb79b65e1a4bf9140cb : SUCCESS in 58m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/69da78f968764181932bd7b6e330322b : SUCCESS in 51m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/08f08d0656dc400abf54a7982deae449 : SUCCESS in 57m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2cd02112938d4e468d62b5df958f8ddf : SUCCESS in 31m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf12f6b1566b4ba29e582e05be701330 : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":1},{"id":"fe2553aece41c46af5e8d43974deb688cdb9788a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-02 08:53:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2c18cad57b8bc6cf7e53f82b88a3d7e5da3df16b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-05-02 11:28:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"268c93bd6fe1ea59c2b5ef6b7f468a7ef97ba53e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 11:29:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3c6fc06241eea129ece58069935353324a62f1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 13:27:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8f726325bba740f1851e7007137dd920 : SUCCESS in 3m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4e64f9e9a6214b4bb972ff1530dfdcbf : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b13b17c7c6854b929f2db19544036166 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3692d43925042329548ba16a8724971 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98c75f3fc22244d7bac0a63411b41253 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d14fa1d0174b44a9a1f086e878eb82b6 : SUCCESS in 4m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0941cf78fb584067a6e83373bf855a1c : SUCCESS in 1h 31m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29e5c16984b841f5aadd3bc554f0abdf : SUCCESS in 3m 52s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/05d181aeb63f4624b3246e65f4cd8dc1 : SUCCESS in 59m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/06d94f90f1c9480eb94b11a9667b5a12 : SUCCESS in 1h 38m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/67bc36641da34f5389092dc79a08dc98 : SUCCESS in 1h 56m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3f004a1f4b6c4da6bc089087aca32491 : SUCCESS in 1h 01m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1087099e5cb94ee0809472ec487523bc : SUCCESS in 1h 35m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/73bc65855ca04a7fa840eb5e8363b33e : SUCCESS in 1h 02m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ba45c17a72ccda1359f347651a3405f15b39d16d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 13:27:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"86d976509b5aafe2df09549ea2324ba3c1f02e8e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 13:29:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/b9b794a4b65349eea45e74a7e5d67145 : SUCCESS in 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6eb3baef42b40ab91894f980b1fbf7b : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9e2393fda2f49568f755989d14964cb : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"daaa1e79b53af6ef601c056cb2030091f7bb9c88","revisions":{"daaa1e79b53af6ef601c056cb2030091f7bb9c88":{"kind":"REWORK","_number":1,"created":"2022-05-01 01:29:48.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/99/840099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/99/840099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/99/840099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/99/840099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/99/840099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/99/840099/1"}}},"commit":{"parents":[{"commit":"1ba863de3d92447448ed817223d7008f192cd558","subject":"Merge \"Refactor session \"is_active\" handling for sqlalchemy-20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1ba863de3d92447448ed817223d7008f192cd558"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 01:09:08.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 01:09:08.000000000","tz":-300},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I67453b5a0e6d41d5cb4ef0cbb47f955a448005eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/daaa1e79b53af6ef601c056cb2030091f7bb9c88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/daaa1e79b53af6ef601c056cb2030091f7bb9c88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
