)]}'
{"id":"openstack%2Fneutron-lib~840720","triplet_id":"openstack%2Fneutron-lib~master~I428b5829e676d4a535044dcfdc1a12bdae77a8f1","project":"openstack/neutron-lib","branch":"master","topic":"wip","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-05-06 10:56:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I428b5829e676d4a535044dcfdc1a12bdae77a8f1","subject":"WIP: test master","status":"ABANDONED","created":"2022-05-05 18:55:09.000000000","updated":"2022-05-06 10:56:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"89d2d4f4ba6884ccd4bd314e6c32a0b5ee5b94ad","_number":840720,"virtual_id_number":840720,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-05 20:16:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 20:16:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0adc4bf6279070da96236eefa65738fc92c6e9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-05 18:55:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2063f7d21c1471a80293c5930f836913a241eec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 20:16:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44f918d15f0a4396925858145c6baf4c : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02d8e3d862b545ad8718406fdf4823a7 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa4cd429225408bad491c71c6525a6f : FAILURE in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a0449abe0b748e7b751a0e464bba0f5 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32df4aa493a846a7a902ef4ca1ab2f0e : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41fc4e3550d44f6c973a8b792afa68ab : SUCCESS in 4m 24s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7513bb5d8fc74f2e92caae7c68a21853 : SUCCESS in 1h 19m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/814420f89b8047eaa36d92462eecf879 : SUCCESS in 30m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e9d333d77ed648cf8baeaee5a2eb8f52 : SUCCESS in 59m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f8e3fd2b3eaa4e3aa4b141aec6abffda : SUCCESS in 1h 03m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a17090b63898426097dc4a56a7682b5a : SUCCESS in 1h 02m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5c79aa944b01410e80bfb6bd22a30dbd : SUCCESS in 54m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9498cbe6313b481e8fadf522ceeb136d : SUCCESS in 34m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1cab1a54d69c449b87874557f53e251b : SUCCESS in 52m 01s\n\nWarning:\n  Comments left for invalid file neutron_lib/api/extensions.py","accounts_in_message":[],"_revision_number":1},{"id":"89d2d4f4ba6884ccd4bd314e6c32a0b5ee5b94ad","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-06 10:56:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84bd03e139db23f86a03f3203187fe9f8463f5c6","revisions":{"84bd03e139db23f86a03f3203187fe9f8463f5c6":{"kind":"REWORK","_number":1,"created":"2022-05-05 18:55:09.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/20/840720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/840720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/840720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/840720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/840720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/840720/1"}}},"commit":{"parents":[{"commit":"daaa1e79b53af6ef601c056cb2030091f7bb9c88","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/daaa1e79b53af6ef601c056cb2030091f7bb9c88"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-05 18:52:47.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-05 18:52:47.000000000","tz":120},"subject":"WIP: test master","message":"WIP: test master\n\nChange-Id: I428b5829e676d4a535044dcfdc1a12bdae77a8f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/84bd03e139db23f86a03f3203187fe9f8463f5c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/84bd03e139db23f86a03f3203187fe9f8463f5c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
