)]}'
{"id":"openstack%2Fneutron-lib~842076","triplet_id":"openstack%2Fneutron-lib~master~I36951d319ec1f3c4227328f1a7315958520eb43d","project":"openstack/neutron-lib","branch":"master","topic":"unbound_ports_request","attention_set":{},"removed_from_attention_set":{"34728":{"account":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"last_update":"2022-05-17 15:47:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I36951d319ec1f3c4227328f1a7315958520eb43d","subject":"Correct error on unbound ports GET request","status":"MERGED","created":"2022-05-17 10:17:03.000000000","updated":"2022-05-17 15:49:07.000000000","submitted":"2022-05-17 15:47:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842076-unbound_ports_request","meta_rev_id":"d70d282c898c471770820545eb4f75101ac70c59","_number":842076,"virtual_id_number":842076,"owner":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-17 15:47:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":2,"date":"2022-05-17 14:31:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2022-05-17 11:07:17.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2022-05-17 14:31:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-17 11:07:17.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-05-17 11:59:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-17 14:31:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"4dd538a823eceba7b2b4a4c4b3623e3ccc705770","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"date":"2022-05-17 10:17:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c846d8c3016e8ca58806faa36889cadd56d4566","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-17 11:07:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b03f142909bd2afac28c80f0d2695ff0612cdce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 11:59:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/89d5c4e96eda484fb659ecef4400d9cd : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c7fdc312f534a4fad9d932ba53a69da : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cab87f2efce140c18d0ce38525f83f46 : SUCCESS in 4m 12s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b39be33df2524d6989c573190e90bebe : SUCCESS in 40m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/75bc02dc8dcb4024a82dbcb8d3ab4609 : SUCCESS in 1h 20m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ee91ff9ced174a35916a7c588f75a507 : SUCCESS in 1h 11m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5043f37f6b9744289aa10c26270468ab : SUCCESS in 59m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e64a41d60af34257b4ebcbe7cd9be67f : SUCCESS in 55m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1682d1b594a44a7183dc804aaad66447 : SUCCESS in 38m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4040799188d4a36ab38d0935dbeaeee : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":1},{"id":"dc04beb27a0f9fa974c0fa5be468c59fb33a204d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-05-17 14:31:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac50b8b3bcdec2f54add7cf0ef95212ae6ce2bc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 14:31:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e2478dc39dd351ca9784175828f824ad3af764fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 15:47:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5d29a76e47bf4a4e9769d85e504733d2 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71378d83fb044fd487661fe2072f905e : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c09824a8d3c4d08adb81be3b7193116 : SUCCESS in 4m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9350712b7d714953b2cb5bb260e0d58d : SUCCESS in 45m 13s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3ff2512289be4b7c8229eae4923e1d5f : SUCCESS in 55m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/eed0d1e8c316483090648ff2f760306e : SUCCESS in 1h 03m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/205174fb27af4268b28b2b6eb38f4888 : SUCCESS in 1h 14m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac1f7ed614a84ceaba34bf0a5017b1c3 : SUCCESS in 1h 02m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/63bf173e49fd4d8185c5c4870d78362d : SUCCESS in 56m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b580679218d48b42968c10bdbb311e17044c71eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 15:47:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d70d282c898c471770820545eb4f75101ac70c59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 15:49:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0a82f0e1a7f340d5a9d98fcfe69ae472 : SUCCESS in 54s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0421a286b61422e997e276ab3fc8068 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd7b5e4a3f198a08331b734913d52eaa01b6c84c","revisions":{"cd7b5e4a3f198a08331b734913d52eaa01b6c84c":{"kind":"REWORK","_number":1,"created":"2022-05-17 10:17:03.000000000","uploader":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"ref":"refs/changes/76/842076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/76/842076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/842076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/842076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/842076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/76/842076/1"}}},"commit":{"parents":[{"commit":"15f1a04698cb17f1962734d04f1eb12a5e0d256f","subject":"Update python testing as per zed cycle teting runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/15f1a04698cb17f1962734d04f1eb12a5e0d256f"}]}],"author":{"name":"Sharon Koech","email":"skoech@protonmail.ch","date":"2022-05-17 10:12:57.000000000","tz":0},"committer":{"name":"Sharon Koech","email":"skoech@protonmail.ch","date":"2022-05-17 10:15:07.000000000","tz":0},"subject":"Correct error on unbound ports GET request","message":"Correct error on unbound ports GET request\n\nSigned-off-by: Sharon Koech \u003cskoech@protonmail.ch\u003e\nChange-Id: I36951d319ec1f3c4227328f1a7315958520eb43d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cd7b5e4a3f198a08331b734913d52eaa01b6c84c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cd7b5e4a3f198a08331b734913d52eaa01b6c84c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
