)]}'
{"id":"openstack%2Fneutron-lib~848445","triplet_id":"openstack%2Fneutron-lib~master~I940cb6b66a9d79b45fb29db1afdfea1aea621f76","project":"openstack/neutron-lib","branch":"master","topic":"remove_constants","attention_set":{},"removed_from_attention_set":{"32667":{"account":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"last_update":"2022-07-05 08:13:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I940cb6b66a9d79b45fb29db1afdfea1aea621f76","subject":"Remove unused placement constants","status":"MERGED","created":"2022-07-01 13:22:34.000000000","updated":"2022-07-05 08:15:21.000000000","submitted":"2022-07-05 08:13:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"848445-remove_constants","meta_rev_id":"60f788c6891d17b87a28a7923272e28f5b89f022","_number":848445,"virtual_id_number":848445,"owner":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-05 08:13:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-07-04 14:14:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2022-07-04 15:41:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-05 06:38:57.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-05 06:38:57.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-01 15:15:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-04 07:27:47.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-07-04 12:57:04.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-07-04 14:14:25.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-07-04 15:41:51.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"}],"messages":[{"id":"e4380030d53a18fb64792c43c07fe7cf19bba7a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2022-07-01 13:22:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"494561c563cafabe783062b4ddde7d05b1d086b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 15:15:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff1b6c02bc96478b85f5aea1ae35b92d : SUCCESS in 6m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca3425660f8d4272b24481c9a0c710b5 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89892228256845088356550342224552 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ce175930b97480bbaed645ef0be9ff2 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/148810065b234d319ba8110416d9194c : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11ef1ba1287f4718abd55bd0565066ca : SUCCESS in 3m 55s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/669ef53add9942b9a9d60a7764d72710 : SUCCESS in 1h 51m 06s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/2ebd542907884af1a43466e4dd72eae1 : SUCCESS in 1h 16m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/435d13f3edc04cbcaa8f46c06d8977cf : SUCCESS in 1h 26m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c63f7cfc6895443b9bb6449a6c0e85fc : SUCCESS in 1h 15m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c6acacc93e8c466f8005d10493598bd0 : SUCCESS in 1h 25m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/af1124a4b81d42858bb478d2fce3040b : SUCCESS in 32m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a20e492d90340eeb59c1a8f0754acfa : SUCCESS in 51m 35s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":1},{"id":"87dbe8c66f3ced861a7e864bd54a5fe267c3e026","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-07-04 07:27:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7eccf2a59239aedab7f110f26bd354fed192b3c9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-04 12:57:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad0861545f0928e56d27b59d6759cdfe77541ca1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2022-07-04 13:31:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d3af08717f941745168d79d32ee3cbf0a069690","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-04 14:14:25.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b459332945a73c199d46d260d075a41d6ce2b11","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-04 14:15:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4486c4b3e8f1e308cd4ecd0d48e1baa36565d4ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 15:33:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39669e11cf744abfb1817db174e5d83e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3c61dc6aed4cf2b5e721d14558bb29 : SUCCESS in 7m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05bb3466c7c1444c94fdc26437c2d072 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f347eae124734965938287174c351f93 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f7e152419fc4106a933c056f88c90dc : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c685aa5fe79415892ac8fb82a3bce36 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e72948c891b4c4993e0b712d6ffd816 : SUCCESS in 4m 08s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07c4fa80c11e40f9b332bfbcc91977f8 : SUCCESS in 1h 23m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f261bf3c2d5457f9684d53217fac35c : SUCCESS in 3m 03s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/94ecebc35d6e4af88ceb603b2c24435a : SUCCESS in 1h 28m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0ee4c4429cec4270824ad78a284261ff : SUCCESS in 1h 17m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/587af9414c5544538f074c78b3146b0d : SUCCESS in 1h 29m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3fce379eb3cf4db4826438abbe82c058 : SUCCESS in 1h 58m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bad548798f9d40b6a5666eb299e1eec9 : SUCCESS in 35m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/16b1a2483f264b2f841682991f9ce882 : SUCCESS in 1h 00m 43s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":2},{"id":"053cab0d1192e96bbe9334a7257d75d196e404f5","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-07-04 15:41:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a5005387a45cc4cbe946bac6a66ab4225a6d7803","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-07-05 06:38:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f9084aa884bec7f1fc097ed7dca999a4cb349b8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 06:39:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5bdc72a5e543deb6a3dbd530e03cee7a3e7a8964","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 08:13:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24c0e08424d5465b809d7227be10faa4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b7cf0ea04d4ad48e02cb0540c2a41d : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aec3ad9d2b1647bd9592203919cb6259 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51e19de72ac44ef5b7149188de99cc42 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53d5e301735d42eb8b340200b0115c0d : SUCCESS in 4m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eeca8e2d102b4ff399d20330cf77f70d : SUCCESS in 1h 29m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bba98f6cc14d42ad89e6863679f601b9 : SUCCESS in 3m 03s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0c07f308cd0a4ddd9feed524408deee8 : SUCCESS in 1h 23m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/21330002097a4dcd8642fdf3486aab14 : SUCCESS in 1h 28m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/572a97922ad94122a19e582ab8747c00 : SUCCESS in 1h 03m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2baa7f04552c461098bbeff305fce9af : SUCCESS in 1h 05m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/24bc329dcee74801b004a614b0ac8fb9 : SUCCESS in 1h 03m 01s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":2},{"id":"0a45a7c11f6622e0bff76d3bbb729a03c2421fbc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 08:13:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"60f788c6891d17b87a28a7923272e28f5b89f022","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 08:15:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a4fe096766d409d9c17d977f8803f28\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/758231de4d7e4774af4ec81fa23ea23b : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6cd30df139a46ddab99255760dea24b : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"89dc650220244628ea78f58ea2771a2ca0c63a73","revisions":{"aed9b89096cf3cd83f1321c6f1cf85ab9e8b48c7":{"kind":"REWORK","_number":1,"created":"2022-07-01 13:22:34.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/45/848445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/45/848445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/848445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/848445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/848445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/45/848445/1"}}},"commit":{"parents":[{"commit":"c8132ff1b06b0d9c50faa19d30f0346cb6a2eea9","subject":"Merge \"Revert \"Introduce new bgp_associations API definition\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c8132ff1b06b0d9c50faa19d30f0346cb6a2eea9"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2022-07-01 13:17:08.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2022-07-01 13:17:08.000000000","tz":180},"subject":"Remove unused placement constants","message":"Remove unused placement constants\n\nNeutron now uses os-resources-classes lib constants [1],\nneutron-lib constants can be removed.\n\n1) https://review.opendev.org/c/openstack/neutron/+/799676/\n\nChange-Id: I940cb6b66a9d79b45fb29db1afdfea1aea621f76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/aed9b89096cf3cd83f1321c6f1cf85ab9e8b48c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/aed9b89096cf3cd83f1321c6f1cf85ab9e8b48c7"}]},"branch":"refs/heads/master"},"89dc650220244628ea78f58ea2771a2ca0c63a73":{"kind":"REWORK","_number":2,"created":"2022-07-04 13:31:33.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/45/848445/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/45/848445/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/848445/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/848445/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/848445/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/45/848445/2"}}},"commit":{"parents":[{"commit":"c8132ff1b06b0d9c50faa19d30f0346cb6a2eea9","subject":"Merge \"Revert \"Introduce new bgp_associations API definition\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c8132ff1b06b0d9c50faa19d30f0346cb6a2eea9"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2022-07-01 13:17:08.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2022-07-04 13:29:56.000000000","tz":180},"subject":"Remove unused placement constants","message":"Remove unused placement constants\n\nNeutron now uses os-resources-classes lib constants [1],\nneutron-lib constants can be removed.\n\n1) https://review.opendev.org/c/openstack/neutron/+/799676/\n\nChange-Id: I940cb6b66a9d79b45fb29db1afdfea1aea621f76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/89dc650220244628ea78f58ea2771a2ca0c63a73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/89dc650220244628ea78f58ea2771a2ca0c63a73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
