)]}'
{"id":"openstack%2Fneutron-lib~849040","triplet_id":"openstack%2Fneutron-lib~master~I41491e7d85647df94ceffda4ccb1d469ecf1ac4a","project":"openstack/neutron-lib","branch":"master","topic":"cascade_delete_extension","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-07-15 18:44:54.000000000","reason":"Change was abandoned"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-07-15 18:44:54.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-07-15 18:44:54.000000000","reason":"Change was abandoned"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2022-07-15 18:44:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I41491e7d85647df94ceffda4ccb1d469ecf1ac4a","subject":"Network Cascade Deletion Extension","status":"ABANDONED","created":"2022-07-07 19:58:36.000000000","updated":"2022-07-15 18:44:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"96596fe75ac8f04300b9879091d89110eed035ce","_number":849040,"virtual_id_number":849040,"owner":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-07 20:43:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}]},"reviewer_updates":[{"updated":"2022-07-07 20:15:58.000000000","updated_by":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-07-07 20:15:58.000000000","updated_by":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2022-07-07 20:15:58.000000000","updated_by":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-07-07 20:43:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-07 22:08:36.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"49de73081cff72988f7f7a80203e1c856d874c45","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"date":"2022-07-07 19:58:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12f75d66f41a65dd90d104ac8d31992f8f3be47b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"date":"2022-07-07 20:17:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0d8f24b9b77d02ffacc910a711bbf35afcab6db9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 20:43:58.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/566ad73a9e674f9fb58308dff36d2036\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97a5046c6c0c4584b15416731e5edfd0 : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc34febb7b9340f080e744c0b75ffc45 : FAILURE in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85289e1a916b449abea5b60d688a2262 : FAILURE in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a55448b86805452cb9d82cee3b690848 : FAILURE in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e56ffe764214f1e8a8e1e190ca96b1c : FAILURE in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b8686f51a9f404e8725ff71997f3725 : FAILURE in 3m 33s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f5b4df213bf4a1fb3eb103adc0f251d : FAILURE in 14m 16s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/771c0898fe59471c9379d48735d9e17c : FAILURE in 14m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5bc6e3f8b0c84796845a41dd377e290f : FAILURE in 13m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9200252cebdd4dd38cdb65ccb930957c : FAILURE in 14m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6975610264454348b6f278934f7adc69 : FAILURE in 15m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/df0a7f91e8f94b18a5a7fad76a00ec18 : FAILURE in 14m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/850242f20b6e463f8878079934ac3520 : FAILURE in 21m 21s","accounts_in_message":[],"_revision_number":2},{"id":"96596fe75ac8f04300b9879091d89110eed035ce","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"date":"2022-07-15 18:44:54.000000000","message":"Abandoned\n\nNo longer needed.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5f4648b661e8b0b6079be42866595a7b5fdbf9f5","revisions":{"54e229fab1bbbd81c024c91f0767e839861f518d":{"kind":"REWORK","_number":1,"created":"2022-07-07 19:58:36.000000000","uploader":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"ref":"refs/changes/40/849040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/40/849040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/849040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/849040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/849040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/40/849040/1"}}},"commit":{"parents":[{"commit":"cd7b5e4a3f198a08331b734913d52eaa01b6c84c","subject":"Correct error on unbound ports GET request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cd7b5e4a3f198a08331b734913d52eaa01b6c84c"}]}],"author":{"name":"Sharon Koech","email":"skoech@protonmail.ch","date":"2022-06-17 11:21:40.000000000","tz":0},"committer":{"name":"Sharon Koech","email":"skoech@protonmail.ch","date":"2022-06-23 14:39:51.000000000","tz":0},"subject":"Network Cascade Deletion Extension","message":"Network Cascade Deletion Extension\n\nChange-Id: I41491e7d85647df94ceffda4ccb1d469ecf1ac4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/54e229fab1bbbd81c024c91f0767e839861f518d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/54e229fab1bbbd81c024c91f0767e839861f518d"}]},"branch":"refs/heads/master"},"5f4648b661e8b0b6079be42866595a7b5fdbf9f5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-07 20:17:34.000000000","uploader":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"ref":"refs/changes/40/849040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/40/849040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/849040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/849040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/849040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/40/849040/2"}}},"commit":{"parents":[{"commit":"89dc650220244628ea78f58ea2771a2ca0c63a73","subject":"Remove unused placement constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/89dc650220244628ea78f58ea2771a2ca0c63a73"}]}],"author":{"name":"Sharon Koech","email":"skoech@protonmail.ch","date":"2022-06-17 11:21:40.000000000","tz":0},"committer":{"name":"Sharon Koech","email":"skoech@protonmail.ch","date":"2022-07-07 20:17:34.000000000","tz":0},"subject":"Network Cascade Deletion Extension","message":"Network Cascade Deletion Extension\n\nChange-Id: I41491e7d85647df94ceffda4ccb1d469ecf1ac4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5f4648b661e8b0b6079be42866595a7b5fdbf9f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5f4648b661e8b0b6079be42866595a7b5fdbf9f5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
