)]}'
{"id":"openstack%2Fneutron-lib~849778","triplet_id":"openstack%2Fneutron-lib~master~Ib51826951aa79fafdd077bd89151e3f5f3a707c7","project":"openstack/neutron-lib","branch":"master","topic":"cascade_delete_extension","hashtags":[],"change_id":"Ib51826951aa79fafdd077bd89151e3f5f3a707c7","subject":"Remove wrong Change-Id from commit message","status":"ABANDONED","created":"2022-07-14 04:48:32.000000000","updated":"2022-07-15 14:06:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"0d7b32e88b514235c3112cbcfb6b554fd0c1cf0d","_number":849778,"virtual_id_number":849778,"owner":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-14 05:11:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8159d8fa426114f10d9a0a8ce703b3b5f8504fb5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"date":"2022-07-14 04:48:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a81eac71f1ad8fe2f20abf12061fced5987d2836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 05:11:55.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2614d3c0a4fa48a3abc4a0a939a58705\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e676d6d147b74668aaed78af0d9c6eb0 : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14a68a8539c7429ba488620546dd97a2 : FAILURE in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0848d39c013743c9ae762982c44f749c : FAILURE in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/329c55c426644b4491287a54bc319d80 : FAILURE in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10c9c9b7890947878dcd397b9767c497 : FAILURE in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37750b8e1cc447b5abd69a2a3484a355 : FAILURE in 4m 08s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a391ed63975545cfbe2075905279d452 : FAILURE in 21m 24s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/7f2c6dea915440ac962af1febf9f2655 : FAILURE in 17m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6a3e77c5f1134cc0885dafb85806ea48 : FAILURE in 17m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1876ba73ff244b028065b7f3339033df : FAILURE in 18m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/79bfb71bdbc648d2b6d80283eb971cbc : FAILURE in 19m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ecd74480ca8a4f0194826660bc07eeb6 : FAILURE in 20m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e399dea2f7894983b2895554c5a6a7ba : FAILURE in 15m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0c1af172028b0358b92621d5e50da9e133e22bcc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"date":"2022-07-15 14:03:13.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0d7b32e88b514235c3112cbcfb6b554fd0c1cf0d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"date":"2022-07-15 14:06:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"27a5ec1f5e0a9c8a0caad3f9706d81847d0421ac","revisions":{"3c45ffef0cd044c7bca85d49e504be5fa72c7b6d":{"kind":"REWORK","_number":1,"created":"2022-07-14 04:48:32.000000000","uploader":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"ref":"refs/changes/78/849778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/849778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/849778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/849778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/849778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/849778/1"}}},"commit":{"parents":[{"commit":"89dc650220244628ea78f58ea2771a2ca0c63a73","subject":"Remove unused placement constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/89dc650220244628ea78f58ea2771a2ca0c63a73"}]}],"author":{"name":"Sharon Koech","email":"skoech@protonmail.ch","date":"2022-06-17 11:21:40.000000000","tz":0},"committer":{"name":"skoech","email":"skoech@protonmail.ch","date":"2022-07-14 03:44:49.000000000","tz":0},"subject":"Network Cascade Deletion Extension","message":"Network Cascade Deletion Extension\n\nChange-Id: I41491e7d85647df94ceffda4ccb1d469ecf1ac4a\n\nNeutron spec: https://review.opendev.org/c/openstack/neutron-specs/+/810822\n\nChange-Id: Ib51826951aa79fafdd077bd89151e3f5f3a707c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3c45ffef0cd044c7bca85d49e504be5fa72c7b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3c45ffef0cd044c7bca85d49e504be5fa72c7b6d"}]},"branch":"refs/heads/master"},"27a5ec1f5e0a9c8a0caad3f9706d81847d0421ac":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-15 14:03:13.000000000","uploader":{"_account_id":34728,"name":"Sharon Koech","display_name":"skoech","email":"sharon.koech@canonical.com","username":"skoech"},"ref":"refs/changes/78/849778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/849778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/849778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/849778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/849778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/849778/2"}}},"commit":{"parents":[{"commit":"89dc650220244628ea78f58ea2771a2ca0c63a73","subject":"Remove unused placement constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/89dc650220244628ea78f58ea2771a2ca0c63a73"}]}],"author":{"name":"Sharon Koech","email":"skoech@protonmail.ch","date":"2022-06-17 11:21:40.000000000","tz":0},"committer":{"name":"skoech","email":"skoech@protonmail.ch","date":"2022-07-15 14:00:07.000000000","tz":0},"subject":"Remove wrong Change-Id from commit message","message":"Remove wrong Change-Id from commit message\n\nSpec: https://review.opendev.org/c/openstack/neutron-specs/+/810822\n\nChange-Id: Ib51826951aa79fafdd077bd89151e3f5f3a707c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/27a5ec1f5e0a9c8a0caad3f9706d81847d0421ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/27a5ec1f5e0a9c8a0caad3f9706d81847d0421ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
