)]}'
{"id":"openstack%2Fneutron-lib~855053","triplet_id":"openstack%2Fneutron-lib~master~I7894c2b5afa11bcb0c78b3525d74bd9345edd361","project":"openstack/neutron-lib","branch":"master","topic":"bp/custom-floatingip-distributed","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-03 09:31:25.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-02-05 17:14:58.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-02-05 19:20:05.000000000","reason":"Change was submitted"},"30380":{"account":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"last_update":"2023-02-05 19:20:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7894c2b5afa11bcb0c78b3525d74bd9345edd361","subject":"[ovn]Floating IP adds distributed attributes","status":"MERGED","created":"2022-08-30 02:40:06.000000000","updated":"2023-02-05 19:21:14.000000000","submitted":"2023-02-05 19:20:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"855053-bp/custom-floatingip-distributed","meta_rev_id":"e79c395cc1e00d91b1787731b50c31ccda9c521d","_number":855053,"virtual_id_number":855053,"owner":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-05 19:20:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-02-03 09:31:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-02-05 17:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2023-02-05 17:14:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-30 04:17:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 07:58:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-09-07 13:03:49.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-09-08 22:07:58.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-09-26 09:46:15.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-10-26 11:30:42.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"93348d7ebf2d2fbf688e711aa78a0e1438594d38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-08-30 02:40:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"760538446ad8b07c21fc35ff2ade574038e50a70","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-08-30 02:40:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"e913d7cd1a8ca9f80e11306a79191f0422352a2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 04:17:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/447403dcdb9f475eb13575008fb03f79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13fc548bb9544e1db171903f42bbdbf5 : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e29c2738a18a4a668057ea09b0b8fe83 : FAILURE in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b2e175b80de498ab3eb36484b128c3c : FAILURE in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/673c26ce998345abb6ae2243504a5658 : FAILURE in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21d0ae491c91440c957df44efa68ece7 : FAILURE in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13627f21699b4ac79142845e75ce2875 : FAILURE in 5m 04s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd441d25f7604609bd0f135ea0ab1da9 : SUCCESS in 1h 28m 53s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/2bef5e59bfa54771b8b7cfc556074f58 : SUCCESS in 1h 29m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2ef226ebf56f414881119fc3d14f6c10 : SUCCESS in 1h 32m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/459939df94684d66b583299f4c4c7a73 : SUCCESS in 1h 23m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8f34b873da0e4308ba7e05bb3c7114cd : SUCCESS in 1h 20m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/22b57ed784514cb99edf51ab68bddc90 : SUCCESS in 36m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf9f23a7a6cb4d429b1ebd1402865bac : SUCCESS in 1h 05m 06s\n\nWarning:\n  Comments left for invalid file .pylintrc\n  Comments left for invalid file neutron_lib/callbacks/registry.py\n  Comments left for invalid file neutron_lib/db/resource_extend.py","accounts_in_message":[],"_revision_number":1},{"id":"ec31daee8a523c4f275bfdc39da746e8673f835a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-08-30 11:27:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c8c256e543fd8ca36bceedd1bc6311a443c6a1ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 13:07:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf931158ee62498caa589abf351c9f1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597c1520e2944e0ba8ec9655981b6e9a : SUCCESS in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a283cbc697a4073b8339d536095a7d4 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6056ea2f217541ceb1774d5f1cbca7e5 : FAILURE in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02bc4ffb120140d6bbf892de1692500b : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f8ec41b451e4d8782ac065ad491dd28 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d4980a4b5be4b0a85b6a5a780c12854 : SUCCESS in 5m 01s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fc7ef4e7efd4060bfcd310132ea737d : SUCCESS in 1h 32m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e06c8d1da91414d87ce729dfe52d47a : SUCCESS in 3m 30s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/af908af53f5f4d0ba8ff8ce494ef1084 : SUCCESS in 1h 17m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fac2de4a84394fa8927bb02888e3c229 : SUCCESS in 1h 27m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0b2a4e1462ca4d40b04daf8ddcc92adb : SUCCESS in 1h 20m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ba5f6a58a5dd4d66aadaa168fc79f390 : SUCCESS in 1h 18m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b0eeeaf98fb64157b9f335e7af711858 : SUCCESS in 37m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4442228a54540e0afc6f636de6c3b55 : SUCCESS in 1h 23m 23s\n\nWarning:\n  Comments left for invalid file .pylintrc\n  Comments left for invalid file neutron_lib/callbacks/registry.py\n  Comments left for invalid file neutron_lib/db/resource_extend.py","accounts_in_message":[],"_revision_number":2},{"id":"a9c22be0d88d2a5a44f0a9e97b28ad8e06b1511a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-09-06 02:34:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c99b78ac4a47880d3344da19e1ea5884a8c589d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 04:34:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ad145be1a104bc098f05fd6b1059b7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e559c6ecd5d44d798edcabe938e89cd : SUCCESS in 5m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a8949a7c95f454481382aba11be025d : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd493f3585f64a268c3d37de6d9852ca : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0416d427932148d994e9bba0fc2412cc : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aac1f3f04991468e9e81870009a0a6fb : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5bbe5b15b2f4cf3afb60e9ece048079 : SUCCESS in 4m 13s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b86a360c169c4340a5d3d77c92adc9db : SUCCESS in 1h 35m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/582d437ba90b41faabf6cc3a77633011 : SUCCESS in 3m 07s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e20ad28406e745dc9757df015d0c86f0 : SUCCESS in 1h 53m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cf110f842f3c44cd91cf914768339253 : SUCCESS in 1h 27m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9ed8bc47f8954916858721b900727476 : SUCCESS in 1h 20m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1413babd5bba42cd9e62d8b98678a409 : SUCCESS in 1h 23m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b3aa3ded2d6b49ebad41ab7e2d949bf3 : SUCCESS in 39m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0126ca45df9144cd80621d8afb6eab28 : SUCCESS in 54m 34s","accounts_in_message":[],"_revision_number":3},{"id":"6bcdc535f135ce31bc65ca85483e3543e6d507f7","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-09-06 05:51:31.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"23cb11a34be10ff65554290df6ca0af0502d81fd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-07 13:03:49.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"58c720612320a72e935e99ca6daeedda7641d05c","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-09-08 22:07:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ae036ebbc619b31dab64d9a21db0fab01c4f2cf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 22:08:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6533e54ec425fd6744bad8689d1c06939033dc78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 23:46:25.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bfa0173c4dc4a29b0d3200b565c028c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f433b6e158824e3fbd9505db96812021 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d641680833e4176be3f4dcaac25d53d : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a30e7365fa8b479a9a8fc22bf225e1df : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9023e7266b2d45f49cadc794ed92a4b4 : SUCCESS in 4m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/830915bb7bc04ea18e1c83f2e1e8b115 : SUCCESS in 1h 22m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d207be92451e44d19601e45b6cbac253 : SUCCESS in 3m 34s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/fd6eb13484ed44b1b28e0f30b4d4e708 : SUCCESS in 1h 33m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8944a6e23d864678bac97ff8f199bd88 : SUCCESS in 1h 36m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9b2258524b9e40d8b043cafb082a32c8 : SUCCESS in 1h 34m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ce26b687e15b4b1799bd2622c61fe3c4 : SUCCESS in 1h 26m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1443ea13a5504571a4363c3dcb09b233 : FAILURE in 53m 58s","accounts_in_message":[],"_revision_number":3},{"id":"6d0e10c1385c99b4ec16928b074467f92d27abd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-09-09 01:20:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6563e5a4a355fe03db2e0e4c0ce9ce3622bc0ea1","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-09-09 01:20:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e4819965bc313d0b7d39e25707ec9688743d611e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 03:25:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1d1c42bc39641d4bdfecf4e72554809\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac461cf7ea1b4a4bbdbda537950afe2a : SUCCESS in 7m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/badd13e246c7411f9190a2ca7d5343d2 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ef87199ff94465b452e793ec6fbd75 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9f56bf04754444a92cca95df2b26e13 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66ed290d92af412191dbd3b7fa8efaea : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94d5ab26ac204368b53ac5a5a7dde838 : SUCCESS in 4m 43s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/968d2491a0a04e4a8873883ea067689b : SUCCESS in 1h 33m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d336fc1017334cb0b6626c55f567b9a6 : SUCCESS in 4m 16s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9d129cce7cfc443e8ca7940c045bc17b : SUCCESS in 1h 42m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/872f3017667548dd92aed0b798a902b9 : SUCCESS in 2h 02m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/df11679ea23041dc95281577fe62126e : SUCCESS in 1h 45m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9306716d7a74436a885417d70ea2e942 : SUCCESS in 1h 34m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c265e446983144a0b03b67064fabfcf4 : SUCCESS in 46m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed6a0b4a7329478b8fbe843f3eb500b9 : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":4},{"id":"fcec8272682adcd4eadbdfa728a77467380fda85","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-13 14:56:26.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ccbe8acb084c48d49bb6ab0a8b6e4f1617e686bb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-13 15:17:19.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5cabd88b4ee59565bd7386dfff87fe10141ce748","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-09-14 00:39:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5f63a6ab170c325ef60b77533a7846ec36ee6cdd","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-09-14 00:44:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1d3d96d0256a5870e99d51c1b517e5b47fb3797f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-09-14 00:46:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"68fc139aca286a74666cd58621e0f90f8c252ca8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 02:17:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d51fff35345c473ea178b1c1b2970286\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93010c12d9214e448bb7a13b6816049d : SUCCESS in 7m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d9f19a5e74841c4ac8270159c6cba06 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bce076540ea4e30a2af8f531601cfef : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e0f04692fc94527a42039dca1b5de50 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a2f746419fb4fb29ef6e07827595a53 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d526016254343878947fbb57ab78a3f : SUCCESS in 4m 35s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/960ce5c80385463881cfac26cfcb968b : SUCCESS in 1h 30m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26dbf31b6b8b4890962c0a378162684b : SUCCESS in 4m 27s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b931420e6fa444b59fd1730fa4529314 : SUCCESS in 1h 17m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bb7517491bbf4a789f205ab3ea5f0a6e : SUCCESS in 1h 27m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac1d6716261b4b599d8f3af535b9e6e8 : SUCCESS in 1h 09m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/08d72dc729304299805fe215e2460bdc : SUCCESS in 1h 08m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5a74734252af4d3d9f912cb2a2ef9889 : SUCCESS in 41m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/96cfc59035ce4421813a592601037bf6 : SUCCESS in 50m 25s","accounts_in_message":[],"_revision_number":6},{"id":"ee2ad524d035265cd9bfc6240d358f077aa85e42","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-26 09:46:15.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3c477a104441f2118c7bdb10303f6be92567c3fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-10-01 07:16:06.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2f6fba0bcd57c7b2c6125d6deba28d5a148b68f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-01 09:18:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e4de13534d24dce9f3beb2dbcefa263\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0de75e4df25f429f8df7955065bde561 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569cb6b966d14960a46e99d51d2fefe0 : SUCCESS in 7m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd6554fbeb2647c897a2f137c0ce3432 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59c0d028b6434acbb4909063a71c0d41 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b429bada9884d169b2d7f1fe302a9ff : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c64020dc2ff47459a5c2002ca83e479 : SUCCESS in 5m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa6c941ac867442f8447ef946c5d40e3 : SUCCESS in 1h 29m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53b0fe0f5dab470289fc8fa745c10dca : SUCCESS in 4m 14s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/8ed480feb49d4489b9752efd9180bb44 : SUCCESS in 1h 56m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6224af8f858b49b0a5f1d411ce3c999f : SUCCESS in 1h 41m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8a1226512e7543cb9cd4c92ebb5446d7 : SUCCESS in 1h 34m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/002eceaa270e4b2a90a6d0a16207899b : SUCCESS in 1h 31m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/798c91dab85d483988bc86ffd5589642 : SUCCESS in 41m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3269e4b489a249e283e3780f2c82c425 : SUCCESS in 48m 16s","accounts_in_message":[],"_revision_number":7},{"id":"f973d37a08d02d5787c308e194e13ba19ce4a3af","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-10-02 06:23:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bf8998eb096bb8a9ae711a59f7027a3ea88374f3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-10-10 09:14:12.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"29e9673e5c2ea43617cd35fb4bb8a1e9d5076f82","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-10-11 01:12:10.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ff9741e47dc544089d241ad49bc326158022a15b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-10-11 06:06:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"61184c4113c168a5d89cd4d0ce39ab17a6d7568e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-10-11 06:08:08.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5d058783ea29e28dc307bfb019168269f0c77d68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 07:41:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01d8db4887ee4827ae701e258f89ec99\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9ad4914c10364743aca125903d14a456 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be066dbbd45549f9a94419d90c2579eb : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8534bfb885d049a09da6c532f47305cd : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6625bfcbf4d74cd0b60116b938605436 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1198e7161eb94c1cbdc996000871d121 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bfae46f65574c08a1379bc55f1abc3c : SUCCESS in 3m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5423f96cbab45dba11af422e1b848b6 : SUCCESS in 1h 32m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e497ef8f1184b5c87a778737771a447 : SUCCESS in 3m 13s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0edaf172561f42458acfbe11d38cc68a : SUCCESS in 1h 28m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/38160068fde349f9a8f6205b50a0d05d : SUCCESS in 1h 27m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/37ea29c6df7b4e719769bdc4ba9c8f47 : SUCCESS in 1h 22m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/63230408493a40909869e79f967a0d49 : SUCCESS in 1h 21m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5beffa8a2efa4e628de5feec80830ddb : SUCCESS in 38m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9184ee1547d847649fd1c8d0ea4b07ba : SUCCESS in 46m 58s","accounts_in_message":[],"_revision_number":9},{"id":"9c6a5d5addbe4bebbea1cf891e61f497edf997a1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-10-25 08:38:10.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6758b4fbcf612abcf8885a54d10abe2703cf9682","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-10-26 11:30:42.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"b6500404c640939bb54474b668a95171369742bb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-10-26 12:57:09.000000000","message":"Patch Set 9: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"161ad7f1b4d3bd345c8e6d50c15ad9ab27b31255","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-08 10:47:04.000000000","message":"Patch Set 9: -Code-Review","accounts_in_message":[],"_revision_number":9},{"id":"d22474d3a876dc0b4a8e38fd43184bdbfb00d20a","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-11-30 02:11:53.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"49705014d2e2ed4e646a47ba4bfe143d9008bef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-12-21 07:06:43.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c402826fefff7260c075be94c0ee88a5a5840af9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-12-21 07:09:39.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f219037adc512b6fd25e0ccc3f5937027cefef0a","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-12-21 07:10:40.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"774a9c454a29df966ff13ccbdc1cb7cc75c81774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 08:40:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6518654f3e04ddcac39d9f24dc69022\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c5694fda92964dd6b5ffc57ee1674343 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/591025f7fee84a15b8db565622a65bbb : SUCCESS in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d9edcb1a74947cca7d513c49721fa8d : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/debd1931b70548fa869c820e6edd5756 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce4a339be2b446c896375f08facec238 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91f457ba435449a0acf6b2b94f03c6ab : SUCCESS in 3m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d9d113deb0742e3a997c37e0fd3515b : SUCCESS in 1h 27m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41e37fe434fc4b2e9f42ad7060ed1f7d : SUCCESS in 2m 58s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0089fbdc8f1b4f39a3a308b00528516b : SUCCESS in 1h 09m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8662de04208240749bd8558197a5b78e : SUCCESS in 1h 19m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/22823fc492b14351a65564cc9c6c526b : SUCCESS in 1h 05m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bd9c9a105a194bb3a8d2887a7e0868e3 : SUCCESS in 1h 14m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/93087ad36d23499e91e4f064708e6878 : SUCCESS in 28m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8143455b9124d1097c5d81e0883fe58 : SUCCESS in 51m 30s","accounts_in_message":[],"_revision_number":11},{"id":"d7ee26db5b2a716fe41b7ebafdb296af8e58499f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-21 17:05:54.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"bb3d4e242d32320bd7fbaccab20396ccfc85b9e9","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-12-22 06:04:03.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"59168ea470d23ee21b45ba2fdac069fe202c14a1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-03 10:44:57.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6236ab7f3b592e8a5569c84090f32b23ab92a774","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-01-04 02:12:16.000000000","message":"Uploaded patch set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"cd6b1adaf6a0b5f813c4ace5812f611e763651dd","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-01-04 02:13:11.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8edcb7850b88e7925bcbd4f9bad1e4b2553fdd57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 04:01:26.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4f2b54f4e814cf68bb8b9b5e4288f0b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e9f6ae2658cd4eca82238bf25d1f4735 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27992011b0a84d2281dcd8f12ab6249a : SUCCESS in 6m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76153db51223427ab00b250e6b5f9eec : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc56af5aea1474e839bb13b320e1f96 : FAILURE in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaa2acf9ce0148c19661ff1857b58d78 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72573020dba244a2b317c5538409ad29 : SUCCESS in 4m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cdbbaed66bff4778a2c73dbd281e0893 : SUCCESS in 1h 43m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe71aab00cc34ff49def7afa58b4bf2d : SUCCESS in 2m 47s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/28ea4caf4cb34f11b5af1afafe130fe4 : FAILURE in 26m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a697439551634687bdeaf78508be6985 : SUCCESS in 1h 01m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c31c01324cc04344a59e05965767f9eb : SUCCESS in 56m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/73d7698b29144f8fb1842de62418b0a8 : SUCCESS in 58m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/83cc18cac956465e85d30585d7b819f1 : SUCCESS in 30m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d5a7846827f414b8f7bad67c08cf322 : SUCCESS in 55m 12s","accounts_in_message":[],"_revision_number":12},{"id":"8fa32949f908fea1c0cfac6f52fc2dd7582c0bd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-01-04 05:54:49.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"dcdcec611ecf51a68ad41f8733af42330016d249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 07:48:04.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4e299e4e6204dc49b4def6cebb3a031\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/93694440b67541e297d0f9cc85369327 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d6d69e9cd2442128144b4453db343ef : SUCCESS in 5m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84b33697d9fc448a9a4cf87f73725346 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbaea020c5f0414faf650e72857b8542 : FAILURE in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d33429b11d14266bc032783f9c92978 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3226ac7c22fe4d6394348ed69fa463af : SUCCESS in 4m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8acd427c8004e9ea9b51a4ab25bd491 : SUCCESS in 1h 47m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14133e769142413a9ae7f89f74ed6e4e : SUCCESS in 3m 06s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/74d1ced7a6034c098ea3eda7fd80220b : SUCCESS in 1h 19m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/71e7010a97514fd3a66df386e253f0fb : SUCCESS in 1h 37m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/87bebaa8d8364529a071bf5f900ec44a : SUCCESS in 1h 34m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c0d853a2071a41e282281838e767e6f4 : SUCCESS in 1h 23m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/469bcb49487f4acc999a47b79b7ee435 : SUCCESS in 26m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bba913a96c7e43608b372a1d1e25045c : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":13},{"id":"3dbe955f60d5261536828cab7af79373684257ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-01-05 02:13:31.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"9eee76fe859afdc721e26662056a0506863bea41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 03:54:06.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c5201ff0a054ee58c3c7866ce9571fe\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2886d260f9634347b5d94828f1d036e0 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f23f4aa82d406e9a30c2d5427e6cf2 : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a04ab678674044ac8af59e5bfedcf90c : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4e30540802043bd862d21a904315bd3 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0ede3399ac34fd0bf198b287c1cc027 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5c65e8dbd2a49459a714f3506360060 : SUCCESS in 3m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ace239344444a1988309733b063c52c : SUCCESS in 1h 33m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5e208cca45346d0941abe76dc5dcbd7 : SUCCESS in 2m 55s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f9573f352aff4e11ae4f6097cc37b25c : SUCCESS in 1h 08m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/87c43e4791734ae7855c9ccda8a4a3ec : SUCCESS in 1h 08m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9de285f327a7407db1acbf1f48193d9a : SUCCESS in 56m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9a9cb9f7ec5a4f2eb36bc569baf15490 : SUCCESS in 1h 03m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0099da9e25b34e1f953fdaf94ecb5202 : SUCCESS in 32m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5947c5e3af694f3eb4d6ca61b97fb75b : FAILURE in 44m 51s","accounts_in_message":[],"_revision_number":14},{"id":"a8b6085f869372f71cc78e05371e3bcaa8e587c0","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-01-05 05:56:59.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e587237d9ae2791134074d2d37c5616119c775f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 07:52:19.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a58caf663a6840dd9eac6e75a055fdb9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d5b741d516e6477db576b059acdc7fc8 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/825c19fd7f814f8eaab3dcee4b991c40 : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4c9e21ae7434482a4fd88e740979925 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29c6df01a282437ca18caee2a6af2a35 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4455dd3c29a4cccb19f130116459cab : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c03174fd527445c2b951362c0f2825da : SUCCESS in 3m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80a12f09e9de4d2fa1239c70164f5a5c : SUCCESS in 1h 49m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afdda6f7df594d5c9e51d39ceb3ed5ac : SUCCESS in 2m 48s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f8e3de4abbb248979280de0d706d89e0 : SUCCESS in 1h 28m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/500fc12a5e1e4bc4a879da7447a259e6 : SUCCESS in 1h 27m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3725f5286b5d427a802fd75967ae44ce : SUCCESS in 1h 33m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ae996bce162e4174b95b1fa965a01fdb : SUCCESS in 1h 36m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cbe58fd79c7544ce89f7774d130e85bf : SUCCESS in 36m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c888c198c0a44229a1c985b610202ecc : SUCCESS in 55m 25s","accounts_in_message":[],"_revision_number":14},{"id":"c4c7b36add65c0eab08486da21386f27f6c9f180","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-03 09:31:25.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"f4acbb71f051276df6d884922308a1f2873b5315","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-05 17:14:58.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"f0fcac75f23f84f1e482fa0eb4d9f24e28ef3ffc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-05 17:15:36.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"b1792c651ddab558aee8c29f07af8bb90c705f0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-05 19:20:04.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd2ab38f595745099f4eb5762819e1ce\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f2a4cd98603446a085c831d1a2d9cc53 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dfc97cc6b36480a9b9d2baf18055155 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5495feb1a5b47e88389f1d90081f658 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/222bf79771fa4073a8f024df95b79d7c : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09323dcc452f4f29ab54399ab7c5482c : SUCCESS in 3m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfab9b6a12364e2e9b1c565d83841a83 : SUCCESS in 1h 58m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe7e5f5775344be79ac37b08b9a6ae44 : SUCCESS in 2m 42s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/7a14205f695f45dc812a3ff6dcd04af0 : SUCCESS in 1h 25m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c08561037ac540a690a59c7aa3448ab7 : SUCCESS in 1h 30m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a0d0da3fa9a64d36afee910371f752c5 : SUCCESS in 1h 12m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d632c5dd998f4483a7ce893c8c9afe09 : SUCCESS in 1h 13m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/13af985d11374ce1a27e9ecb4a54251a : SUCCESS in 1h 10m 55s","accounts_in_message":[],"_revision_number":14},{"id":"b93599da2b08644c1a48432c3fbefb5995471ef6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-05 19:20:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"e79c395cc1e00d91b1787731b50c31ccda9c521d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-05 19:21:14.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f306403a8dfd40a6b68c52d1c3c942c7\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/f7cd214b5e1347bca8e32f028eb78c41 : SUCCESS in 46s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a8c7cf12680410fa53443b0b8f12317 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce52fefd36444219b9b74be5962c3b4b : SUCCESS in 42s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"7441b091ee13b22376fbfb970b5372d9d6b4a00c","revisions":{"b234f7c3c8d5bb618b87422678ac26f77336c351":{"kind":"REWORK","_number":1,"created":"2022-08-30 02:40:06.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/1"}}},"commit":{"parents":[{"commit":"e8f0d31d38d397f148191d89c3b83907c688b514","subject":"Merge \"Add floating-ip-port-forwarding-detail shim extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e8f0d31d38d397f148191d89c3b83907c688b514"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:47.000000000","tz":480},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b234f7c3c8d5bb618b87422678ac26f77336c351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b234f7c3c8d5bb618b87422678ac26f77336c351"}]},"branch":"refs/heads/master"},"845ea64c704894d4cc0f94e6fa398bbbd6ba5227":{"kind":"REWORK","_number":2,"created":"2022-08-30 11:27:00.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/2"}}},"commit":{"parents":[{"commit":"e8f0d31d38d397f148191d89c3b83907c688b514","subject":"Merge \"Add floating-ip-port-forwarding-detail shim extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e8f0d31d38d397f148191d89c3b83907c688b514"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 11:26:35.000000000","tz":480},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/845ea64c704894d4cc0f94e6fa398bbbd6ba5227"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/845ea64c704894d4cc0f94e6fa398bbbd6ba5227"}]},"branch":"refs/heads/master"},"6348e62a082409dfe505ba49101fc127e14d73f0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-06 02:34:47.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/3"}}},"commit":{"parents":[{"commit":"7026199065fc67977871c6f2521ce6334c20147a","subject":"Fix pep8 job issues with pylint\u003d\u003d2.15.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7026199065fc67977871c6f2521ce6334c20147a"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2022-09-06 02:34:47.000000000","tz":0},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6348e62a082409dfe505ba49101fc127e14d73f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6348e62a082409dfe505ba49101fc127e14d73f0"}]},"branch":"refs/heads/master","description":"Rebase"},"de2aca3da63b8c7faad41824347af0ff9f670a32":{"kind":"REWORK","_number":4,"created":"2022-09-09 01:20:11.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/4"}}},"commit":{"parents":[{"commit":"7026199065fc67977871c6f2521ce6334c20147a","subject":"Fix pep8 job issues with pylint\u003d\u003d2.15.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7026199065fc67977871c6f2521ce6334c20147a"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-09-09 01:19:41.000000000","tz":480},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/de2aca3da63b8c7faad41824347af0ff9f670a32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/de2aca3da63b8c7faad41824347af0ff9f670a32"}]},"branch":"refs/heads/master"},"a8301955f89da65de910017fe37286cbec867dc6":{"kind":"REWORK","_number":5,"created":"2022-09-14 00:39:45.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/5"}}},"commit":{"parents":[{"commit":"eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed","subject":"Merge \"Add DSCP mark 44\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-09-14 00:39:38.000000000","tz":480},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a8301955f89da65de910017fe37286cbec867dc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a8301955f89da65de910017fe37286cbec867dc6"}]},"branch":"refs/heads/master"},"b1cf254cf0b6c3e7c95e90c8fa84f6f622dcdebc":{"kind":"REWORK","_number":6,"created":"2022-09-14 00:46:01.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/6"}}},"commit":{"parents":[{"commit":"eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed","subject":"Merge \"Add DSCP mark 44\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-09-14 00:45:40.000000000","tz":480},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b1cf254cf0b6c3e7c95e90c8fa84f6f622dcdebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b1cf254cf0b6c3e7c95e90c8fa84f6f622dcdebc"}]},"branch":"refs/heads/master"},"97a113c657db6be4c6b54b3594bfa020d9b059ec":{"kind":"REWORK","_number":7,"created":"2022-10-01 07:16:06.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/7"}}},"commit":{"parents":[{"commit":"eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed","subject":"Merge \"Add DSCP mark 44\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-10-01 07:15:10.000000000","tz":480},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/97a113c657db6be4c6b54b3594bfa020d9b059ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/97a113c657db6be4c6b54b3594bfa020d9b059ec"}]},"branch":"refs/heads/master"},"7cad179e8a79aefa42abf22d9affa7af132032a5":{"kind":"REWORK","_number":8,"created":"2022-10-11 06:06:59.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/8"}}},"commit":{"parents":[{"commit":"eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed","subject":"Merge \"Add DSCP mark 44\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-10-11 06:06:27.000000000","tz":480},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7cad179e8a79aefa42abf22d9affa7af132032a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7cad179e8a79aefa42abf22d9affa7af132032a5"}]},"branch":"refs/heads/master"},"5760dedd4520ea64137fe8f6a88b7a9103c2e4b4":{"kind":"REWORK","_number":9,"created":"2022-10-11 06:08:08.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/9"}}},"commit":{"parents":[{"commit":"eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed","subject":"Merge \"Add DSCP mark 44\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-10-11 06:07:51.000000000","tz":480},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5760dedd4520ea64137fe8f6a88b7a9103c2e4b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5760dedd4520ea64137fe8f6a88b7a9103c2e4b4"}]},"branch":"refs/heads/master"},"bbd1b2ac1a34da92943c787ad75e71d65843eb38":{"kind":"REWORK","_number":10,"created":"2022-12-21 07:06:43.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/10"}}},"commit":{"parents":[{"commit":"eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed","subject":"Merge \"Add DSCP mark 44\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-12-21 07:06:22.000000000","tz":480},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bbd1b2ac1a34da92943c787ad75e71d65843eb38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bbd1b2ac1a34da92943c787ad75e71d65843eb38"}]},"branch":"refs/heads/master"},"4caa03bae743ddda29914a80c289daf8410b7c31":{"kind":"REWORK","_number":11,"created":"2022-12-21 07:09:39.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/11"}}},"commit":{"parents":[{"commit":"eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed","subject":"Merge \"Add DSCP mark 44\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-12-21 07:09:25.000000000","tz":480},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4caa03bae743ddda29914a80c289daf8410b7c31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4caa03bae743ddda29914a80c289daf8410b7c31"}]},"branch":"refs/heads/master"},"be6598e57dca35aa7434ac18d6c3006a257b84db":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2023-01-04 02:12:16.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/12"}}},"commit":{"parents":[{"commit":"eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed","subject":"Merge \"Add DSCP mark 44\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eb4da30f2629100b4c0fc9a6572aedf2fb34b8ed"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-01-04 02:10:19.000000000","tz":480},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartial-bug: #1978039\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/be6598e57dca35aa7434ac18d6c3006a257b84db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/be6598e57dca35aa7434ac18d6c3006a257b84db"}]},"branch":"refs/heads/master"},"c1862082f0d911f7189a28971560b869a989cabf":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-01-04 05:54:49.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/13"}}},"commit":{"parents":[{"commit":"f2bb7c72cd1fa8bfd25431834dde6dbfcf4d0b72","subject":"\"convert_to_sanitized_mac_address\" accepts netaddr.EUI type values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f2bb7c72cd1fa8bfd25431834dde6dbfcf4d0b72"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2023-01-04 05:54:49.000000000","tz":0},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartial-bug: #1978039\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c1862082f0d911f7189a28971560b869a989cabf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c1862082f0d911f7189a28971560b869a989cabf"}]},"branch":"refs/heads/master","description":"Rebase"},"7441b091ee13b22376fbfb970b5372d9d6b4a00c":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-01-05 02:13:31.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/53/855053/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/855053/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/855053/14"}}},"commit":{"parents":[{"commit":"7e9f66aca498706747b23e01d7e0622c7c187002","subject":"Fix a few broken links in api-ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e9f66aca498706747b23e01d7e0622c7c187002"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-30 02:36:38.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2023-01-05 02:13:31.000000000","tz":0},"subject":"[ovn]Floating IP adds distributed attributes","message":"[ovn]Floating IP adds distributed attributes\n\nPartial-bug: #1978039\n\nPartially-Implements: blueprint custom-floatingip-distributed\n\nChange-Id: I7894c2b5afa11bcb0c78b3525d74bd9345edd361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7441b091ee13b22376fbfb970b5372d9d6b4a00c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7441b091ee13b22376fbfb970b5372d9d6b4a00c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
