)]}'
{"id":"openstack%2Fneutron-lib~855737","triplet_id":"openstack%2Fneutron-lib~master~Ib52809e68accca8d086c38e93c086bba01c34152","project":"openstack/neutron-lib","branch":"master","topic":"release-health-check","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-07 09:15:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib52809e68accca8d086c38e93c086bba01c34152","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2022-09-03 20:31:31.000000000","updated":"2022-09-07 09:15:09.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b02eca3587d0d8bab17d2dbb4304e34658cb1a08","_number":855737,"virtual_id_number":855737,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-05 10:08:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-03 22:37:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-05 08:10:38.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"}],"messages":[{"id":"be260fbf6699b4e4bae0c0097e174f7c4b889b18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-03 20:31:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea9d249a1f60ed61983e708cf048a7eacac1688e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 22:37:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04e349c7e3774583a264bb2b52695964\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c53b45ce6ba24772b5d583cb2596af94 : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b85e7b6ce893479cbe43050dd83750ef : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f550621a924250ad90e7e976203a63 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dc3a08708c1435dacc758501003ab28 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acf7fc5e406847a0a063cace7d173004 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d4d230691654f75b255bde6e769d4f1 : SUCCESS in 4m 13s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0913c23f021c408298e8b83c844784b8 : FAILURE in 1h 32m 17s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/bda2beee8fac437291ca554c7ad2477f : SUCCESS in 1h 48m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/de90fa1ed8df454a8fba4bb084a69a2a : SUCCESS in 1h 57m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/64c1bc4ef28d418abcda9225c2fd4ca3 : SUCCESS in 1h 45m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3ee3628e1b5b4240854987f76d00a0f3 : SUCCESS in 1h 42m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6a083f86dfee4ec0b2ca934de1d5757a : SUCCESS in 53m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b9acbd65bee74f1f856b6115b6177c8f : SUCCESS in 1h 06m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0712ce0d64430fedb24436b3621008dabb13baa4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-05 08:10:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb4f23a1d8b61612c0b608c266899238465a1d84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 10:08:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99ef35a70ba2455ab206336019935bcb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40dcb8adb3941e6b4e75c7b6367a455 : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6f1d388b095460897285757f1c6acbf : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97dae1276ca5421b91acd008e348bf7a : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/492077e345134edc9121b9dcfba75710 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e8f19851cc4464d8881cb1c963a63e0 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b0eaceb0ebd4fd8a76d790701500a59 : SUCCESS in 3m 40s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8b9b0edc12d446f83a75f903538183e : SUCCESS in 1h 44m 16s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/8dcd534546524324b7c62bba9ae4335a : SUCCESS in 1h 32m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1f70c57abf02446db9e402f2f6a5c489 : SUCCESS in 1h 51m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/323a44e08c1c4354806186891a1e6d63 : SUCCESS in 1h 39m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b338e1060dad40b69ba5ab5e5b9ce33b : SUCCESS in 1h 33m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/195c1072deae4a7891a641b13a592605 : SUCCESS in 49m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a95ed2d70a341cab0e45ac27d48e738 : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b02eca3587d0d8bab17d2dbb4304e34658cb1a08","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-07 09:15:09.000000000","message":"Abandoned\n\nGate check passed, no further need for this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c7506b140423f7a5d08fe06be156e7d131b5989","revisions":{"5c7506b140423f7a5d08fe06be156e7d131b5989":{"kind":"REWORK","_number":1,"created":"2022-09-03 20:31:31.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/37/855737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/37/855737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/855737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/855737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/855737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/37/855737/1"}}},"commit":{"parents":[{"commit":"7026199065fc67977871c6f2521ce6334c20147a","subject":"Fix pep8 job issues with pylint\u003d\u003d2.15.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7026199065fc67977871c6f2521ce6334c20147a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:31:17.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:31:28.000000000","tz":120},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: Ib52809e68accca8d086c38e93c086bba01c34152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5c7506b140423f7a5d08fe06be156e7d131b5989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5c7506b140423f7a5d08fe06be156e7d131b5989"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
