)]}'
{"id":"openstack%2Fneutron-lib~856740","triplet_id":"openstack%2Fneutron-lib~stable%2Fzed~I0cbafde3d59c3094af214c9cabc846a12947a129","project":"openstack/neutron-lib","branch":"stable/zed","topic":"create-zed","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-09-27 12:51:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0cbafde3d59c3094af214c9cabc846a12947a129","subject":"Update .gitreview for stable/zed","status":"MERGED","created":"2022-09-09 11:47:38.000000000","updated":"2022-09-27 12:52:06.000000000","submitted":"2022-09-27 12:51:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856740-create-zed","meta_rev_id":"f2353f8fbf2b63ed105acb96d3cda1d97ad0ae29","_number":856740,"virtual_id_number":856740,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-27 12:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-09-27 10:51:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-09-27 09:08:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2022-09-27 09:20:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-09-27 09:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-09-27 10:51:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}],"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-09 12:00:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-10 13:39:04.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2022-09-27 08:54:41.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"CC"},{"updated":"2022-09-27 09:20:50.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-09-27 09:24:11.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-09-27 10:51:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"81eaebbd496fbfdf320073b99bb8ee23f558899e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-09-09 11:47:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"125e530a5a9a260c6a904f710fd759b8d309266a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 12:00:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/687d4c666775455ca25d4d3d4f461059\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e26cd13b2644a4a5c3e32a1143f54c : SUCCESS in 5m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af0cc3e47a7b4dcb9e76b73a2197e1bd : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83c0d899074a48eaa991ef4cf302e2ee : FAILURE in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c5100a110b94ca1b93a0eaae449a37d : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1805dbe94b884ec093c4f03ef4c07ecf : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fbae7c24c8741daa9e5df1fe0fdce9a : SUCCESS in 4m 39s (non-voting)\n\nWarning:\n  Comments left for invalid file .pylintrc\n  Comments left for invalid file neutron_lib/callbacks/registry.py\n  Comments left for invalid file neutron_lib/db/resource_extend.py","accounts_in_message":[],"_revision_number":1},{"id":"a3a3545548d522f9dd9546625e4ead77524e508d","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-09-10 13:39:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"293ea670392171b56f0eb7ba97d5b4fae12d61d7","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2022-09-27 08:54:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a14a1d7fe738bfffc99d79048d7b4c2fadb2a84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-09-27 09:08:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1301ff7fd9123f9dae1533f39136fee195b8e96b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-09-27 09:20:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bc106ed8282849e11af0bcef2524bf346ed7054c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-27 09:24:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"33c3fce84a7f3ff9410ca0e021946b2a91503069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 10:51:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21b2e8bfdddd4da8b70982df4c91e3c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dabee4d3933049bba2a40b7c7e4833a2 : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dffe29eba0774c50bccffc9f1da92b26 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d683049a70e64983b21b7819c273945d : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96603dc1330d4c678fd7971ce712897c : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a805cc09a7524cd2b217ca219b4b112a : SUCCESS in 6m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9110e2cce40d4126b2c5f13cae695f56 : SUCCESS in 4m 33s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bb59c8c692847b281e72b1876cc3725 : SUCCESS in 1h 40m 07s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c1e8e6e6107c44cb824117830652a243 : SUCCESS in 1h 28m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7e99c8b922d44b1fbb3c983d677bb6c7 : SUCCESS in 1h 22m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f343696d38314d5d9c1ca152d171a834 : SUCCESS in 1h 06m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6e9ff78590844b9a8bf878be3edf050d : SUCCESS in 1h 18m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/662e0181b76a4c7fa0ea3d3a554af54d : SUCCESS in 31m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f63d5c9cedcb452889448e9dca6e0b97 : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":2},{"id":"0a70eecfd73ad0b09ccac6ea09eaff7b29f81174","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-27 10:51:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d09a2ddbdd9c71abb961d0b641110affc60227df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 10:52:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1904965f690df4fe72cbb2ed54788a48a17d8134","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 12:51:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/909c530bed5448edb189cd649b834ee7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2369edd69244cd9b09fa094acbaf340 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4ee33d2a2594d59a584089375510997 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8c1444123334c17a55207c2a89a3988 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7bd24e6544741a6a4a4ea21dec26ac9 : SUCCESS in 4m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f3f42f126fc423e926cc83f8ff13efa : SUCCESS in 1h 23m 35s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/8b137d83a1b544dfbd8d1519fb684c0e : SUCCESS in 1h 57m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2597242d850848f6ba32595a0ae210ce : SUCCESS in 1h 31m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ae3c943feb49471db921dc38e4cb4dbd : SUCCESS in 1h 06m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7bd6a4f83ad24a8ca57723095478db06 : SUCCESS in 1h 19m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ebc722fa62824ed1858f4dd4585fd5ef : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":2},{"id":"8af0b4a2dcebd7ebca2dd32a446f050a07a15e09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 12:51:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f2353f8fbf2b63ed105acb96d3cda1d97ad0ae29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 12:52:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a123e03f90e46f99d29e64acb102fba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff56242762db4d88a6eda090d492f915 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c30d950669b922460a80b97af075a433e90da45f","revisions":{"3280feaac8d837400387f47c094ec51453b3fdb7":{"kind":"REWORK","_number":1,"created":"2022-09-09 11:47:38.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/40/856740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/40/856740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/856740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/856740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/856740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/40/856740/1"}}},"commit":{"parents":[{"commit":"e8f0d31d38d397f148191d89c3b83907c688b514","subject":"Merge \"Add floating-ip-port-forwarding-detail shim extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e8f0d31d38d397f148191d89c3b83907c688b514"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-09 11:47:37.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-09 11:47:37.000000000","tz":0},"subject":"Update .gitreview for stable/zed","message":"Update .gitreview for stable/zed\n\nChange-Id: I0cbafde3d59c3094af214c9cabc846a12947a129\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3280feaac8d837400387f47c094ec51453b3fdb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3280feaac8d837400387f47c094ec51453b3fdb7"}]},"branch":"refs/heads/stable/zed"},"c30d950669b922460a80b97af075a433e90da45f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-09-27 09:08:52.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/40/856740/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/40/856740/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/856740/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/856740/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/40/856740/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/40/856740/2"}}},"commit":{"parents":[{"commit":"c92a574c357aaefb1fe9089355ab948d7485a7a3","subject":"Fix pep8 job issues with pylint\u003d\u003d2.15.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c92a574c357aaefb1fe9089355ab948d7485a7a3"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-09 11:47:37.000000000","tz":0},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-09-27 09:08:52.000000000","tz":0},"subject":"Update .gitreview for stable/zed","message":"Update .gitreview for stable/zed\n\nChange-Id: I0cbafde3d59c3094af214c9cabc846a12947a129\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c30d950669b922460a80b97af075a433e90da45f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c30d950669b922460a80b97af075a433e90da45f"}]},"branch":"refs/heads/stable/zed","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
