)]}'
{"id":"openstack%2Fneutron-lib~857472","triplet_id":"openstack%2Fneutron-lib~stable%2Fzed~I3a4a27ec4672f8ea8848d7c04651730dae6f40ff","project":"openstack/neutron-lib","branch":"stable/zed","topic":"bug/1988296-stable/zed","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-09-19 10:57:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a4a27ec4672f8ea8848d7c04651730dae6f40ff","subject":"Fix pep8 job issues with pylint\u003d\u003d2.15.0","status":"MERGED","created":"2022-09-13 17:58:44.000000000","updated":"2022-09-19 10:58:08.000000000","submitted":"2022-09-19 10:57:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857472-bug/1988296-stable/zed","cherry_pick_of_change":855356,"cherry_pick_of_patch_set":2,"meta_rev_id":"274c60e8c35ac3bfadfdfaa4ec41d394657ea8a4","_number":857472,"virtual_id_number":857472,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-19 10:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-09-16 08:22:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-09-19 10:45:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-09-15 00:37:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-17 08:25:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-09-19 10:45:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-13 17:58:44.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-09-13 18:13:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-15 00:37:25.000000000","updated_by":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"REVIEWER"},{"updated":"2022-09-16 08:22:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-09-17 08:25:02.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-09-19 10:45:20.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"fed89c1333bd2ca23d5e40b2912de25ef930a806","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-13 17:58:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"22fa67c10d1c4e32c0f4207e7233bd6c396ffd7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 18:13:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b12aa7be000b4a6e914afbe3531f830b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7fcb4e387d4405a413d60d14cf0ff1 : SUCCESS in 7m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/910e58a422494214a2b248958c38311b : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64396bb59ad14037b4370738d1ad15da : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/625ac25fa08144899fe631b84b546313 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c86988fef807475f8ba8ac1ceb92df35 : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1dd00e41cd549b1a54f8ab5bb29045b : SUCCESS in 6m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2b75b09656d38eedd97bfca7ce0e1b72ead41e0","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-09-15 00:37:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b17b6488b4eee8f7d98e98b1728a28af9c4a056d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-16 08:22:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"13679de38bdbf1cfcae4707f742caf60f799f2f2","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-09-17 08:25:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0bd7863e92da4bf7505cdcec356584de4baa55f3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-19 10:45:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e374329d987a419cea0087425f05f9372c11c59b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 10:45:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7bb949082fd0939f49a518fe83c63fb762861809","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 10:57:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c457027eae734a7a83984eedf010400a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ededfd8f544ad1a60b4eb110e6c024 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/642b6096ae3645959223d1a02ff32081 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/992769196063487d9635c884bdd9ab11 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e66167e5de1495e8a958267f816ed77 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1b7066a6bb469bf576d5cfdff364694e0da99fa1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 10:57:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"274c60e8c35ac3bfadfdfaa4ec41d394657ea8a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 10:58:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67635f8745084fcca5785d818e07f496\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d5490af57af4946a372a7e05b9c36d9 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c92a574c357aaefb1fe9089355ab948d7485a7a3","revisions":{"c92a574c357aaefb1fe9089355ab948d7485a7a3":{"kind":"REWORK","_number":1,"created":"2022-09-13 17:58:44.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/857472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/72/857472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/72/857472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/72/857472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/72/857472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/72/857472/1"}}},"commit":{"parents":[{"commit":"e8f0d31d38d397f148191d89c3b83907c688b514","subject":"Merge \"Add floating-ip-port-forwarding-detail shim extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e8f0d31d38d397f148191d89c3b83907c688b514"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-31 10:57:04.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-09-13 17:58:44.000000000","tz":0},"subject":"Fix pep8 job issues with pylint\u003d\u003d2.15.0","message":"Fix pep8 job issues with pylint\u003d\u003d2.15.0\n\nCloses-Bug: #1988296\nChange-Id: I3a4a27ec4672f8ea8848d7c04651730dae6f40ff\n(cherry picked from commit 7026199065fc67977871c6f2521ce6334c20147a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c92a574c357aaefb1fe9089355ab948d7485a7a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c92a574c357aaefb1fe9089355ab948d7485a7a3"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
