)]}'
{"id":"openstack%2Fneutron-lib~858990","triplet_id":"openstack%2Fneutron-lib~master~I779b1d0445df874eca1a59a17d54e00a68f4c865","project":"openstack/neutron-lib","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2022-09-23 21:17:17.000000000","reason":"\u003cGERRIT_ACCOUNT_4694\u003e replied on the change","reason_account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2022-09-26 11:32:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I779b1d0445df874eca1a59a17d54e00a68f4c865","subject":"Fix missed six library items in .pylintrc","status":"MERGED","created":"2022-09-22 21:55:07.000000000","updated":"2022-09-26 11:33:58.000000000","submitted":"2022-09-26 11:32:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858990-remove-six","meta_rev_id":"095fe0435f6e2a767be267e4c09ff3d1e3eb96da","_number":858990,"virtual_id_number":858990,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-26 11:32:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2022-09-23 21:17:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-09-26 09:47:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-25 04:56:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2022-09-24 01:12:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2022-09-26 09:47:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-22 23:37:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-23 11:21:08.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-09-23 17:30:09.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-09-24 01:12:09.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-09-25 04:56:49.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"9946c266b465cd3edd399b8a25b9284269e2d2af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-09-22 21:55:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01c1c692b4c98f0d298a40d52755e3886c95f358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 23:37:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0de6e8d70734cdd970b4e6c99aa5199\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab7c62452b14020a1155516e40e76b7 : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e9ed59c3e9d4ed7b1adf92050e10247 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1453804f9f77436eb090170735f56d50 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d9bac4fea664b87a757d469aed180d8 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ba6b8837cf3443e810e818b0a5a3190 : SUCCESS in 4m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/697611602df846b695de961a7ee334db : SUCCESS in 1h 36m 29s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/007c6f26384d4578937cc13f1b414427 : SUCCESS in 1h 11m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/10489f7330ac47d6b51c8e9e9c855576 : SUCCESS in 1h 11m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/48e0ddf365764b67892b1fc740a4b4c2 : SUCCESS in 58m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/088e9a236bac48c69f60d7c975518418 : SUCCESS in 1h 04m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0924aefd2a1045d2b7f5644bcbeea067 : SUCCESS in 35m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d06ab74bd404d3196383abc935b24a7 : SUCCESS in 1h 02m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fcac0fedc3fc37f397f2613ed933ffbc6d7d53d1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-23 11:21:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0cee57673299aa3b7fa62e73940af502d89801d7","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-09-23 17:30:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38ef347f2c7f2148050d4289fca4a903634c9409","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-09-23 19:21:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"871f891baf030b30b6d1f629eb73d0349098b62a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-09-23 19:21:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"29d6609c9821804b2a5144d41cfe06b75dbc7d8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 21:14:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c36eb33be5e24383a9739facffdcb4d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31f6e408ff4d4cb190fef89f5d539a04 : SUCCESS in 5m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/505767d78c7a449d96b55aa26c155cb3 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f878eb8e862401fbeb958fe8ad16b65 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85f189d9c47b4fc4b588c750027f53d1 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0d93d2411974d19a1030ba8b027cb39 : SUCCESS in 3m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7313c352a3034fe297ba388e4379248e : SUCCESS in 1h 48m 13s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/85a540f389874459830719addd5e8752 : SUCCESS in 1h 18m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e8f69e19374f409796f490b59d53fc23 : SUCCESS in 1h 35m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b52014d1b9044a1ca26140bf1af22826 : SUCCESS in 1h 19m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4db1d1836d1448e0a4f65fc672400e6a : SUCCESS in 1h 17m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/28037daaee0b4a82ad9d098a4dc253e2 : SUCCESS in 40m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c888c61ba4a94239a4329a3c96f68fe6 : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":2},{"id":"23c80305233cae8407893c90ddd9e317b10115de","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-09-23 21:17:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7230bfbe1c73bf3a37eb615fd6fffd581631138e","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-09-24 01:12:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a90300671db643d6a1a169ce940ca92d8a4a89ac","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-09-25 04:56:49.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"82be82b1aa6b5dd6f071366532fd6cdafb53606b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-26 09:47:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"66fc12ecf7fcc7d0f7848f8ed89e3a28f2b70f7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 09:48:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a0005a6a5e555d075ff8cb6c3c620f321204b762","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 11:32:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c433c97f560846e99129ee6468b63c18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfac85dbefb94940b4ca01103525873e : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a27fa26a8974411908be626fca4f860 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1133e1ec36504d499be776b50db7cd70 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7953095f1c564c618ba3338f466f2786 : SUCCESS in 3m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ec149961f354e809d1f420b98e42867 : SUCCESS in 1h 36m 23s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/79422e9928cf4f1181b9c2418202d3e1 : SUCCESS in 1h 43m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/baa335d34ffc4cc89887bc07ae7aa765 : SUCCESS in 1h 42m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/550455b74a7d418ca280417c57a3aa2c : SUCCESS in 1h 32m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1bcfe46bbfbc4531a5951f6e33aa3756 : SUCCESS in 1h 35m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/229b43cafe814fb8a5ad608ef959cc86 : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":2},{"id":"eadf876a0c505257e03bf850ee8ebdab8ab52016","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 11:32:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"095fe0435f6e2a767be267e4c09ff3d1e3eb96da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 11:33:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5517e1a985bd40238b2f9fa5bd4887e6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0701fab250a40f49de1e61db40dda31 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9284ee76fffd151feddb4f68cdb8c0198bacfba","revisions":{"ebed2c055fbc2f3aba18c1ffb1990a3a80290d8b":{"kind":"REWORK","_number":1,"created":"2022-09-22 21:55:07.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/858990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/90/858990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/858990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/858990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/858990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/90/858990/1"}}},"commit":{"parents":[{"commit":"f831de7f2fcf301b7e39069b8b005f50d066700a","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f831de7f2fcf301b7e39069b8b005f50d066700a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-09-22 21:51:56.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-09-22 21:51:56.000000000","tz":-240},"subject":"Fix missed six library items in .pylintrc","message":"Fix missed six library items in .pylintrc\n\nThe six library was removed and depcrecated a while ago,\nbut .pylintrc wasn\u0027t updated accordingly. This makes it\nmatch what was done in the neutron repo.\n\nTrivialfix\n\nChange-Id: I779b1d0445df874eca1a59a17d54e00a68f4c865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ebed2c055fbc2f3aba18c1ffb1990a3a80290d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ebed2c055fbc2f3aba18c1ffb1990a3a80290d8b"}]},"branch":"refs/heads/master"},"d9284ee76fffd151feddb4f68cdb8c0198bacfba":{"kind":"REWORK","_number":2,"created":"2022-09-23 19:21:45.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/858990/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/90/858990/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/858990/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/858990/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/858990/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/90/858990/2"}}},"commit":{"parents":[{"commit":"f831de7f2fcf301b7e39069b8b005f50d066700a","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f831de7f2fcf301b7e39069b8b005f50d066700a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-09-22 21:51:56.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-09-23 19:21:16.000000000","tz":-240},"subject":"Fix missed six library items in .pylintrc","message":"Fix missed six library items in .pylintrc\n\nThe six library was removed and depcrecated a while ago,\nbut .pylintrc wasn\u0027t updated accordingly. This makes it\nmatch what was done in the neutron repo.\n\nTrivialfix\n\nChange-Id: I779b1d0445df874eca1a59a17d54e00a68f4c865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d9284ee76fffd151feddb4f68cdb8c0198bacfba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d9284ee76fffd151feddb4f68cdb8c0198bacfba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
