)]}'
{"id":"openstack%2Fneutron-lib~862296","triplet_id":"openstack%2Fneutron-lib~master~Ia30f679a371911501c4a458b4764da7011535dca","project":"openstack/neutron-lib","branch":"master","topic":"nova-binding-profile","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-11-04 09:17:46.000000000","reason":"Change was submitted"},"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2022-11-04 09:17:46.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-11-03 14:08:57.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-11-04 09:17:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia30f679a371911501c4a458b4764da7011535dca","subject":"Document that binding:profile is for compute services only","status":"MERGED","created":"2022-10-21 12:12:33.000000000","updated":"2022-11-04 09:18:49.000000000","submitted":"2022-11-04 09:17:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"862296-nova-binding-profile","meta_rev_id":"c129f6f29224d4f02a12ff0f1ef84aa32e2d1a14","_number":862296,"virtual_id_number":862296,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-04 09:17:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2022-11-03 14:08:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2022-11-04 07:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-03 16:16:35.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2022-11-04 07:43:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 12:26:28.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-10-21 12:26:28.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-10-21 13:40:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-02 10:05:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-11-02 15:39:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2022-11-03 16:16:35.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"8a21b6c24c88e6b46b91dc13d83b355d31512eed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2022-10-21 12:12:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ba23122ea75ff816b0cdec87caf64cb70ac6973","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2022-10-21 12:24:06.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8ce103ddf081f4eab0ef3e5970e35f120541a50d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 13:40:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/360e09f85afb438ea2f481628bb4e804\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f94f4d6b87b34e1ea2c95b286982d857 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a76aa37098b14e2fbbac8d9d6ed7758f : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4584757c314f3f99ed410757beef8e : SUCCESS in 4m 14s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/ec46693e6f3c4484bac92a6085b12e4f : SUCCESS in 1h 23m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/28feaf87183b4ee28bf7d0c29b12ad15 : SUCCESS in 1h 24m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ced43cf39d8e496f9490cf9ed053f2b4 : SUCCESS in 1h 15m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f0bb86432b264b45a649ea030b02299f : SUCCESS in 1h 22m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5d07de5d25434cc5baa6a22aad322de6 : SUCCESS in 32m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4cbaeb0b312f457dbd7a7cc735bb3810 : SUCCESS in 44m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4d4d98c69dba570c957c97b774a6905a0d678e42","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-02 10:05:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"faca96dc4033fa6b2a670c5b585962f8f2d7eafa","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-11-02 15:39:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74db35da9a9e7def8c1af97e50825912d9a35463","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-02 18:01:17.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d7a0724000d8bbda9f9e9d43d7ce7c766474a9b9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-03 09:32:25.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"970a66ec4fe5013695747332e9414e8015178c0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2022-11-03 09:46:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"599241c95d0149e8b107efeaf5d75d8c0d8bd1bd","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2022-11-03 09:48:28.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"956f15354860f25863c314ad043c11509acb9381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 11:06:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e75e1bd58504f46b9a2ad1cb3730512\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4cfb8a9069504d0cb55d7def669c1039 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/446071619bb64737bc2f3c11777b1b0a : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd714b675f57494da961f296c568ffbe : SUCCESS in 3m 43s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/711633574d2249e2a51bdd6bb44b7215 : SUCCESS in 1h 06m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3b53d6c404d34ebbbeed40931d0a815f : SUCCESS in 1h 17m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e6318029832a415186e99e2369ddd506 : SUCCESS in 58m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6b4fffe483f6481fa416e17949fc4939 : SUCCESS in 1h 03m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/734e1638654a451d8469f4aa7725e788 : SUCCESS in 30m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d58e6f8d7439458aa09a1bf1613204af : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":2},{"id":"3b9077faa1f98cd35da3deb4698b1fe3428c6cf5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-03 14:08:57.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c2cab034c426c91a8edae319f183fbbab645c56b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-11-03 16:16:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"002688f0597808a79207c676959e6139cb0e30ca","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-04 07:43:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7063fc65d4ffe580d19a32ede1906fc84548c8d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 07:43:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f09dd38e2a0965689c6a3901b3bf0b35b3d8ccfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 09:17:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a10a9864b3514a3d88f59f7e0015074d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/265813976200487eb49b727ebe58b4d8 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75933bbc477b4084b1b259593c549e80 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b38a1b88caa8491393a6321612bf690c : SUCCESS in 4m 28s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b637f2ef61414c18add982a83b123d71 : SUCCESS in 1h 26m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b0dfcd4af9144df1a91395321c42da12 : SUCCESS in 1h 29m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/53d09afc7e1f45c48fd14ca44b6f69c3 : SUCCESS in 1h 21m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0690bf0ce6b04706a58b398f5451a8f0 : SUCCESS in 1h 15m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d594b835d53e478e80f37dc9b9962a47 : SUCCESS in 39m 33s","accounts_in_message":[],"_revision_number":2},{"id":"589737ceb32b49060876061c19da35a5d08ef9a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 09:17:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c129f6f29224d4f02a12ff0f1ef84aa32e2d1a14","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 09:18:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a81ab54aa0942ebae20cbee50a4318c\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/540e8aabfa3d4630b79edf0c185ac5a9 : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/566368c0a2f148249639628be177ecb5 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6100385c769da489fb982b9bc4bd32bc78f87ff0","revisions":{"f8494886c3cbe43249901fd0f92641dc7e8ecb16":{"kind":"REWORK","_number":1,"created":"2022-10-21 12:12:33.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/96/862296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/96/862296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/862296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/862296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/862296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/96/862296/1"}}},"commit":{"parents":[{"commit":"9ecd5995b6c598cee931087bf13fdd166f404034","subject":"Use oslo_context.from_dict() for context generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9ecd5995b6c598cee931087bf13fdd166f404034"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2022-10-21 11:48:54.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2022-10-21 12:10:51.000000000","tz":120},"subject":"Document that binding:profile is for nova only","message":"Document that binding:profile is for nova only\n\nReferences:\nhttps://meetings.opendev.org/meetings/neutron_drivers/2022/neutron_drivers.2022-09-30-14.01.log.html#l-28\nhttps://meetings.opendev.org/irclogs/%23openstack-neutron/%23openstack-neutron.2022-09-30.log.html#t2022-09-30T16:08:59\n\nCounterexamples:\nhttps://docs.openstack.org/neutron/rocky/admin/config-ovs-offload.html\nhttps://specs.openstack.org/openstack/nova-specs/specs/rocky/implemented/sriov-trusted-vfs.html\n\nChange-Id: Ia30f679a371911501c4a458b4764da7011535dca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f8494886c3cbe43249901fd0f92641dc7e8ecb16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f8494886c3cbe43249901fd0f92641dc7e8ecb16"}]},"branch":"refs/heads/master"},"6100385c769da489fb982b9bc4bd32bc78f87ff0":{"kind":"REWORK","_number":2,"created":"2022-11-03 09:46:06.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/96/862296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/96/862296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/862296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/862296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/862296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/96/862296/2"}}},"commit":{"parents":[{"commit":"30b74827f735bb16669b7ffdaabc2aec002bb77e","subject":"Api-ref: Add note that only VLAN segmentation-id can be changed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/30b74827f735bb16669b7ffdaabc2aec002bb77e"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2022-10-21 11:48:54.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2022-11-03 09:40:30.000000000","tz":60},"subject":"Document that binding:profile is for compute services only","message":"Document that binding:profile is for compute services only\n\nReferences:\nhttps://meetings.opendev.org/meetings/neutron_drivers/2022/neutron_drivers.2022-09-30-14.01.log.html#l-28\nhttps://meetings.opendev.org/irclogs/%23openstack-neutron/%23openstack-neutron.2022-09-30.log.html#t2022-09-30T16:08:59\n\nCounterexamples:\nhttps://docs.openstack.org/neutron/rocky/admin/config-ovs-offload.html\nhttps://specs.openstack.org/openstack/nova-specs/specs/rocky/implemented/sriov-trusted-vfs.html\n\nChange-Id: Ia30f679a371911501c4a458b4764da7011535dca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6100385c769da489fb982b9bc4bd32bc78f87ff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6100385c769da489fb982b9bc4bd32bc78f87ff0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
