)]}'
{"id":"openstack%2Fneutron-lib~862808","triplet_id":"openstack%2Fneutron-lib~master~Id675e897b23bce1b478ab6c1e438c1d253c53a07","project":"openstack/neutron-lib","branch":"master","topic":"bug/1806052","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-10-31 04:41:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id675e897b23bce1b478ab6c1e438c1d253c53a07","subject":"Api-ref: Add note that only VLAN segmentation-id can be changed","status":"MERGED","created":"2022-10-27 12:50:31.000000000","updated":"2022-10-31 04:42:53.000000000","submitted":"2022-10-31 04:41:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862808-bug/1806052","meta_rev_id":"45f25b2fabf16dc05e1e71fb27dd5b33e78ca18c","_number":862808,"virtual_id_number":862808,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-31 04:41:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-10-27 14:01:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-10-31 03:32:18.000000000","_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-10-31 03:32:18.000000000","_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-27 14:01:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-10-27 14:10:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-31 03:32:18.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"}],"messages":[{"id":"14abeab54c1cd0da5173b9fa54d28ae18958c17f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-10-27 12:50:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24dfe082af361c52d1a26fc5be3f8b87e4e62cd1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-10-27 14:01:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6e4283a4c61d7323f28e793e9387fb75ddc7813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 14:10:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48e507efb05444ea93ff42cae8cfaa70\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b03afa8172dc42399d6dffbff2d03df6 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70e180f9364c457b9ca9163d4ffe9d3c : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60fa4fbe3df041748fb2e5473b2de7e2 : SUCCESS in 4m 14s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/2796847b03d84e9aaa3dfb86901522e0 : SUCCESS in 1h 12m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6d2ec97057fc4558b238b808a870fd4e : SUCCESS in 1h 18m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/15f08c3d8f0e47249d0a452378b5161b : SUCCESS in 58m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/139b96c1d1e444fcb574dab1ce9470dc : SUCCESS in 1h 03m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e2f9b0914b9747b889d6745def37195e : SUCCESS in 27m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3dc98af75c284fe0a592d2b4462d2392 : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7419432be92c3c7919c3e82872f915eac7b0e229","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-10-31 03:32:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f6ea2d70f6fe634315e4954f984ea0d04f1483c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 03:32:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"827a8c5590d9a25d97bbf0ed6efe3b0a92d1413e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 04:41:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f337681076834e5fb4528c579d59e4f5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6ed85a79ab004990b2d22ea23e60112d : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d45aafb76b429ab00e4e427f46cba3 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f5b0ba14ab4522837867e2f5b2a2b3 : SUCCESS in 3m 32s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/8dcbc34557bf419e8a5394b3693555ef : SUCCESS in 1h 07m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8e220365bd6a4e25ae9ead4d9f5b8c7e : SUCCESS in 1h 03m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a8c98f4aa4f644a19f77ebb97c224f2e : SUCCESS in 1h 05m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c387a320471e4949b995fb727b7bc417 : SUCCESS in 57m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d25dab806006431ea3ab1d06872dc334 : SUCCESS in 37m 39s","accounts_in_message":[],"_revision_number":1},{"id":"0a0694b4ac0093e71b8f3e4355924f54d66116aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 04:41:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"45f25b2fabf16dc05e1e71fb27dd5b33e78ca18c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 04:42:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cdf0e8fd68242d38b33724910481e9d\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/bf85ccc0bf20485cb393141e0308a36f : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54eae875b2e144d9a6d46c186f75c3c6 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30b74827f735bb16669b7ffdaabc2aec002bb77e","revisions":{"30b74827f735bb16669b7ffdaabc2aec002bb77e":{"kind":"REWORK","_number":1,"created":"2022-10-27 12:50:31.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/08/862808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/08/862808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/862808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/862808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/862808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/08/862808/1"}}},"commit":{"parents":[{"commit":"0964b73b442dbc1a5bded9fbef5922c7e1834255","subject":"Add flows tables for distributed metadata datapath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0964b73b442dbc1a5bded9fbef5922c7e1834255"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-10-27 12:10:27.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-10-27 12:43:41.000000000","tz":120},"subject":"Api-ref: Add note that only VLAN segmentation-id can be changed","message":"Api-ref: Add note that only VLAN segmentation-id can be changed\n\nSince [1] it is allowed to change the provider:segmentation-id of VLAN\ntype networks.\n\n[1]: https://review.opendev.org/q/topic:bug%252F1806052\n\nCloses-Bug: #1428062\nRelated-Bug: #1806052\nChange-Id: Id675e897b23bce1b478ab6c1e438c1d253c53a07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/30b74827f735bb16669b7ffdaabc2aec002bb77e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/30b74827f735bb16669b7ffdaabc2aec002bb77e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
