)]}'
{"id":"openstack%2Fneutron-lib~864568","triplet_id":"openstack%2Fneutron-lib~master~I0531ea2c1bf29aecfab9b491eefef044a4ee49ad","project":"openstack/neutron-lib","branch":"master","topic":"bug/1996606","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-11-16 15:05:36.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-11-16 13:29:01.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-11-16 15:05:36.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-11-16 12:45:12.000000000","reason":"\u003cGERRIT_ACCOUNT_5948\u003e replied on the change","reason_account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}}},"hashtags":[],"change_id":"I0531ea2c1bf29aecfab9b491eefef044a4ee49ad","subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","status":"MERGED","created":"2022-11-15 16:33:22.000000000","updated":"2022-11-16 15:06:56.000000000","submitted":"2022-11-16 15:05:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864568-bug/1996606","meta_rev_id":"600e233b4f7d03d9bdfdb7cb7df971a5737fafa8","_number":864568,"virtual_id_number":864568,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-16 15:05:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-11-16 11:16:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-11-16 13:29:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-11-16 12:45:12.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-11-16 13:29:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-15 17:56:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-16 10:44:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-11-16 10:44:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-11-16 10:44:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-11-16 11:16:33.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"22339a26f82a573323f75219fb7f41ace38d80ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-15 16:33:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c485b3705ef5d53d0bba8b6b2030211459415c9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 17:56:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07bf6ad50dd040a9b80a6c39074ffcdc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/739d64f658024997b9769e5a7fe42d63 : SUCCESS in 5m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29a670902c274a48913a569f74abb9ca : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a29a7106624e23a8f11a2cc9a941cf : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ad76f8adbc34ab1893552d1ca44c163 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fcf3c3814bc4808aa3759805878302e : SUCCESS in 3m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/839dd7f4418f4d128b1abb3d7e3fb760 : SUCCESS in 1h 21m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0cb1ce0c029445dacdecc516cb317e5 : SUCCESS in 3m 13s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/602389383f2242aeb82f2ca00d1f10f9 : SUCCESS in 1h 08m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/99e7907989bd45fc81ac01081ff353cb : SUCCESS in 1h 04m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a7d4145b9724a37ae36c3ab8d4314d6 : SUCCESS in 1h 13m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8fed58866cf54b6cbf3b26021d353441 : SUCCESS in 1h 06m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aa820e8c3ff94b06a99033b4660ca5c4 : SUCCESS in 32m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f0146d2e10c24eca9ecd97900413c307 : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":1},{"id":"01b9f9945a9541598e40c494bbdce592f98311c2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-16 11:16:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b90793534121cdb5e0e15acd84c408b4f8dcb414","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-11-16 12:45:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"65b71bb1e96bec4f2b28537bd9cabdf41897fb8e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-11-16 13:29:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"227348207b2c5d17245f82c6f3a7a9b46470075e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 13:29:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1f7b6650fb4987b41de2cbc484857b336ba2cd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 15:05:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27c8bff47e584c00a290aacf34046fa9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b26153bfee924bbca964a21b1b1e40a1 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d94b88845a0435e8ea940fe5cd0cd49 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77d5156ab76b47afae6d7998cbe72950 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0e9769d97ef4313958470eef2944c89 : SUCCESS in 3m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b0a07b4ccfe4a9f9af6a79409414f94 : SUCCESS in 1h 32m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14324be4b8104cd19fe45b63324e6a33 : SUCCESS in 2m 47s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/2e090191ce82431d82c5a08af1b4a9fe : SUCCESS in 1h 04m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/919c1cb954cd461ab9e27a24757ee531 : SUCCESS in 59m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/30e154a2b5554276b09a0766dd777b2b : SUCCESS in 1h 00m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b9a627af3d7c4d4999ce2ab9c44c3686 : SUCCESS in 56m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7ede79d1acad415d9b2e49b381a3f8c3 : SUCCESS in 50m 10s","accounts_in_message":[],"_revision_number":1},{"id":"de81cf9d0dce9aa12c09986aee0afd474fcfd20e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 15:05:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"600e233b4f7d03d9bdfdb7cb7df971a5737fafa8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 15:06:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9022755380449fbbbfbe58500e822d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1ae63263ad8496c8612438d4ebe87da : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb49b2e25a004a39bd89a8b558006ff6 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e4a6beed43d28f7ab0c6c47ea056c70ea7b1fea","revisions":{"7e4a6beed43d28f7ab0c6c47ea056c70ea7b1fea":{"kind":"REWORK","_number":1,"created":"2022-11-15 16:33:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/68/864568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/68/864568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/864568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/864568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/864568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/68/864568/1"}}},"commit":{"parents":[{"commit":"d9284ee76fffd151feddb4f68cdb8c0198bacfba","subject":"Fix missed six library items in .pylintrc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d9284ee76fffd151feddb4f68cdb8c0198bacfba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 10:14:02.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 10:30:38.000000000","tz":60},"subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","message":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"\n\nThe plugin \"qos\" is added to the \"EXT_PARENT_RESOURCE_MAPPING\" constant.\nThe policy engine can now check the QoS rule ownership using the QoS\npolicy project ID. The QoS rules are resources that do not have an\nassigned project ID, it is inherited from the QoS policy. This patch\nallows to check a QoS rule ownership using the \"ext_parent_policy_id\"\nfield, that stores the QoS policy project ID.\n\nNOTE: once released, this bug fix must include Neutron unit tests\nchecking, using the rules:\n  \"update_policy_bandwidth_limit_rule\": \"rule:admin_or_owner\"\n  \"update_policy_packet_rate_limit_rule\": \"rule:admin_or_owner\"\n  \"update_policy_dscp_marking_rule\": \"rule:admin_or_owner\"\n  \"update_policy_minimum_bandwidth_rule\": \"rule:admin_or_owner\"\n  \"update_policy_minimum_packet_rate_rule\": \"rule:admin_or_owner\"\n\nCloses-Bug: #1996606\nChange-Id: I0531ea2c1bf29aecfab9b491eefef044a4ee49ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e4a6beed43d28f7ab0c6c47ea056c70ea7b1fea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e4a6beed43d28f7ab0c6c47ea056c70ea7b1fea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
