)]}'
{"id":"openstack%2Fneutron-lib~864735","triplet_id":"openstack%2Fneutron-lib~stable%2Fzed~I0531ea2c1bf29aecfab9b491eefef044a4ee49ad","project":"openstack/neutron-lib","branch":"stable/zed","topic":"bug/1996606","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-11-23 09:07:24.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-11-22 14:10:52.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}},"hashtags":[],"change_id":"I0531ea2c1bf29aecfab9b491eefef044a4ee49ad","subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","status":"MERGED","created":"2022-11-16 14:00:31.000000000","updated":"2022-11-23 12:19:11.000000000","submitted":"2022-11-23 12:18:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864735-bug/1996606","meta_rev_id":"0e525270b15eaa250f5ec3adb792f597acb2af1d","_number":864735,"virtual_id_number":864735,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-23 12:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-11-22 14:10:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-11-17 13:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-19 04:49:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-11-22 14:10:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-16 15:56:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-17 13:46:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2022-11-17 13:51:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-11-19 04:49:30.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-11-21 16:41:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"02146a12a59d40ed5915a5e2e43798be83472747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-16 14:00:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16d8c5ba166ba8053ebcb3ded85f99b9ef56d292","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 15:56:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd67a098b37c45728e1ba3e06b958b9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e48b1d900247188d362fbb249da1a4 : SUCCESS in 8m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b153533bd2aa4c2296b09091d4e258dd : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9be2559445994f8699745a3bb8be0615 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f96032edd68e4e8fa39ba6c6aa07c209 : POST_FAILURE in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb2a9b5792e747d38ec351b56be6f546 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2f79ad4e83e465a8717472417657e27 : SUCCESS in 3m 25s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86f43f0230b54c26b3022a583ebf3f8c : SUCCESS in 1h 48m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5311225730345c2a7233bc27a64debf : SUCCESS in 2m 57s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b36fdce53bb740648b69e4475a3f4846 : SUCCESS in 1h 20m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7ca53ae176e546168d869467b9199eaf : SUCCESS in 1h 12m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9b742dee61d34cc88ef6adcebe125333 : SUCCESS in 1h 05m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d4c2357dcdc04f48908567829ee45895 : SUCCESS in 1h 07m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6cdbdb433f8b4741bd349e49e64f3100 : SUCCESS in 26m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fce0aadbefac41f0916883484e57c7f0 : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e986be2e1a96adc4ca8c0571a490c205a24b8ba8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-17 13:24:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1de12c2bf3352e959eeedc5097478190c8f5129","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-17 13:46:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26270f4860ad0b65b5d6506f17f43fb9fd43f3a7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-17 13:51:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dfd436f135e8c3fd77de9cdbd658935142206e89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 14:44:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c91e5b8b5aa49108a81f30f18e297ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce6088974ffa47b08066f9dc7abf515a : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19c7ddd2f5f44d1a9de24ccb6a8916cb : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6df5542666d644b9bb16daadfc37f188 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd697a264c1247d9b0769217b0bed5e3 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c8403f5d7dc4f88af41ea771fbe3644 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48a09b6397c446d093312eb54770650f : SUCCESS in 4m 05s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a07d02c7c6164c188b0e04e04b3d4a28 : SUCCESS in 1h 17m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c03b3dec2f584b63b7a6b7991910aae5 : SUCCESS in 4m 09s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/2cb130c34c2644c1a8257f2429675f7f : SUCCESS in 1h 07m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9ebd1ac825b546b495dd0e18bfaf0a88 : SUCCESS in 1h 05m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f057214c83494cb895c05ac770239a2a : SUCCESS in 1h 06m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/66da0ee5873d4592b4a9ae69a26eabd0 : SUCCESS in 1h 00m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fa0b375fc95643c88fe7aad4dd0cd741 : SUCCESS in 28m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/38224c042a294c67b990621ee20f62ee : SUCCESS in 45m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5778087917cb7246c7d28777c1bfcc4a9777c5d3","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-11-19 04:49:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e76a488454710b6149a5535c2f73dc618de0ba1d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-22 14:10:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"505c37447386aedf8be47c741d3a8eee93476709","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 14:11:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35943b2e0cc3bfc664ca2d2f5a3deb9ac184c695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 16:13:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9be12657c3db4db09f62d80c9c89bb84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6cb64991c8b4efaae04f2b5e48390b0 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/278763b35c5c410ca1957afb06e512f3 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c00a26bf71b462c84b19e555603e09c : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c362ccbae99c4767b632d344f4b108a5 : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f28fa0e7ef734d948052f83b3175007f : SUCCESS in 1h 54m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8203ba03b5dc4c17ab95eac1c47c9878 : SUCCESS in 3m 27s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a5eea6f3c6094e15a171874163a59238 : SUCCESS in 1h 29m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/db2acd2ca8684c6ba91bcd6db6c85bfc : SUCCESS in 1h 18m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b3f31fc175474596a6f094a2d47c4790 : SUCCESS in 1h 11m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/86a214f740a543cca125c7738bc82fa2 : FAILURE in 1h 05m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/58c532af07fb4878bfdbd27eb5fb43a5 : SUCCESS in 46m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d65746e3edb00d936570c226f469bf76a1a6e01b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-23 09:07:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9b867ec20b2da61bb9b7914cfabb1714987e8bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 10:56:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35ec765e4c8e4f04962189b21245fe97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/381fa3d548964f8f8938cb1ebf1d3cc8 : SUCCESS in 4m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fb317c31c6348b6ae0e685728268f7a : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b0b2ab51be4e01a3a49f7e049eb273 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f3675937a094754a1f2689f4cdd3eaf : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d269aa49a74b4a898ddb038b696f0133 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02da117c19744b73990d70aa0d6f7b38 : SUCCESS in 3m 57s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d24349aff6304acba52af8f4c3c72ec2 : SUCCESS in 1h 20m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/349f3e3bc9ec4fa3bb8e9787dc2a60c9 : SUCCESS in 3m 25s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/28454229ca1941c6877d009f3be507db : SUCCESS in 1h 33m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/24b35f4e3be34be1854703a788b881db : SUCCESS in 1h 45m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6dfa713ff1e84836a7068e88caa0a5b7 : SUCCESS in 1h 29m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5bc80430415a4a94842ead801dc5e9a1 : SUCCESS in 1h 33m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/70506a869ba94fd9a9d31e5823bde7df : SUCCESS in 38m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/11e9855f915f46199c4316188c48757b : SUCCESS in 47m 16s","accounts_in_message":[],"_revision_number":1},{"id":"06adebc85bd9e589194d2933d06e066876f92036","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 10:56:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5f788827476a67f9543002aae748a4072654cff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 12:18:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5abc8279c8f144c6bb0a59f84bafc328\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad4ea389d3594f1a979d67938fd8fe69 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8da08320a884fa2a452cfc99688a0cc : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25f29877e11344d0884924d4aae8eb08 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/603675542bb949718e8f693916beb346 : SUCCESS in 3m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8850a38eb1b74363bd67019406ef737b : SUCCESS in 1h 19m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01128070bd8f46ce9e159a7c2a34e590 : SUCCESS in 3m 01s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d40036ed8c9f42d5b77b9a569a961d6d : SUCCESS in 1h 18m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/baaf08825ece4822b79b7f7fc186e590 : SUCCESS in 1h 09m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bc6623c2c3c844c7bbac897975390958 : SUCCESS in 1h 06m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4286171228734a019ea2cb424e350008 : SUCCESS in 56m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3426dd606d04c6fa6ca8d962b54cebd : SUCCESS in 48m 22s","accounts_in_message":[],"_revision_number":1},{"id":"978342ca30a00569fa14d9f9c27f17ad55f9fb98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 12:18:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0e525270b15eaa250f5ec3adb792f597acb2af1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 12:19:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cf6e4c48b3c4b27bdb46b6b938fb328\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/804608c40b7b49ab8c748cb3ada24163 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a77dfb11ed24a85892d946af7d0ad6d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64b45c627ac164ca075542b095d34d80b12c9b95","revisions":{"64b45c627ac164ca075542b095d34d80b12c9b95":{"kind":"REWORK","_number":1,"created":"2022-11-16 14:00:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/35/864735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/35/864735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/864735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/864735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/864735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/35/864735/1"}}},"commit":{"parents":[{"commit":"1db700635dc2ca8cdeaeac48691dfbf0e081373a","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1db700635dc2ca8cdeaeac48691dfbf0e081373a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 10:14:02.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-13 07:57:46.000000000","tz":60},"subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","message":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"\n\nThe plugin \"qos\" is added to the \"EXT_PARENT_RESOURCE_MAPPING\" constant.\nThe policy engine can now check the QoS rule ownership using the QoS\npolicy project ID. The QoS rules are resources that do not have an\nassigned project ID, it is inherited from the QoS policy. This patch\nallows to check a QoS rule ownership using the \"ext_parent_policy_id\"\nfield, that stores the QoS policy project ID.\n\nNOTE: once released, this bug fix must include Neutron unit tests\nchecking, using the rules:\n  \"update_policy_bandwidth_limit_rule\": \"rule:admin_or_owner\"\n  \"update_policy_packet_rate_limit_rule\": \"rule:admin_or_owner\"\n  \"update_policy_dscp_marking_rule\": \"rule:admin_or_owner\"\n  \"update_policy_minimum_bandwidth_rule\": \"rule:admin_or_owner\"\n  \"update_policy_minimum_packet_rate_rule\": \"rule:admin_or_owner\"\n\nCloses-Bug: #1996606\nChange-Id: I0531ea2c1bf29aecfab9b491eefef044a4ee49ad\n(cherry picked from commit 7e4a6beed43d28f7ab0c6c47ea056c70ea7b1fea)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64b45c627ac164ca075542b095d34d80b12c9b95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64b45c627ac164ca075542b095d34d80b12c9b95"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
