)]}'
{"id":"openstack%2Fneutron-lib~864738","triplet_id":"openstack%2Fneutron-lib~stable%2Fxena~I0531ea2c1bf29aecfab9b491eefef044a4ee49ad","project":"openstack/neutron-lib","branch":"stable/xena","topic":"bug/1996606","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-11-24 15:35:12.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"hashtags":[],"change_id":"I0531ea2c1bf29aecfab9b491eefef044a4ee49ad","subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","status":"MERGED","created":"2022-11-16 14:02:57.000000000","updated":"2022-11-24 18:25:49.000000000","submitted":"2022-11-24 18:24:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864738-bug/1996606","meta_rev_id":"f8e4b67bb21f3e2c3c1f6898ef2eca1b1449b2ad","_number":864738,"virtual_id_number":864738,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-24 18:24:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-11-22 14:11:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-11-23 18:31:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-11-23 18:31:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-16 15:26:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-22 14:11:04.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-11-23 18:31:25.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"25599bcd5b3787c7fb2b0d3de4f8b0cc4323fcae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-16 14:02:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06295bfb672f17d63e5fb40556aabc4e15ae2712","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 15:26:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62097a65f38d4785a5b63dda02d09919\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6b01ce016a5439a82a663f1f48e3c3e : SUCCESS in 3m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ab88e2df9b64c5785eafb13b5e94dad : POST_FAILURE in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/154bd882b39148e8b5ef3b641653ac66 : FAILURE in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6507ac38df9a4804a1ecb4aa5ed6470e : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6de7586c1d734e6a92ecc3dc9ed83f25 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f105414f7ff4110932c6bfbf54e7cca : SUCCESS in 3m 42s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fe4bda952684062b2c3bdc7b2335f23 : SUCCESS in 1h 22m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15477291116d4aa0bbb0f89ef3a8fe02 : POST_FAILURE in 4m 36s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/877250b37dc543be847faadce1cf94ac : SUCCESS in 1h 02m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5235811bfe4490e984675ec9010c277 : FAILURE in 45m 22s\n\nWarning:\n  Comments left for invalid file .pylintrc\n  Comments left for invalid file neutron_lib/api/extensions.py\n  Comments left for invalid file neutron_lib/callbacks/registry.py\n  Comments left for invalid file neutron_lib/db/resource_extend.py","accounts_in_message":[],"_revision_number":1},{"id":"e06efaa4a0d2ef2e7af1a7f93b6f424f11af1bcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-17 13:22:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"43784c039701be31863db7d8c61695ea5ab7c414","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 14:55:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/193e23145a4a497383f46c2ee50414db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32f6036f70a041669b0d0e7b49b0a19a : SUCCESS in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e4f9317b4cc455294cd16b030a207a2 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60c1631e674f4704bc1a092cddf295b7 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07bf4229fb2a43be859cc100155dce6f : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81c892e47a8049aa999cfd2502d19bac : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1745bfed5b24a9c8238c2f1eafb2a48 : SUCCESS in 3m 33s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6aecd70a4fa344ffbbcdf92690c075c7 : SUCCESS in 1h 19m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38c7ea38be4d43f08bab66398bff6177 : SUCCESS in 2m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d4eb28de946c415dad3b7cb7a4b2437c : SUCCESS in 1h 20m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bff0946bb10844acbd900466043d8834 : SUCCESS in 45m 49s","accounts_in_message":[],"_revision_number":2},{"id":"df753b77479a372511e8e7db7a6ddd59b534063b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-22 14:11:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"42cc987af57ce27e6959969d9fd5bfe69f9fc48b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-11-23 18:31:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0834a98736fd724a7b1159a9b84594565ed14950","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 18:31:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b85d34fa786849ff6470179cda6c7c86d703c6fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 19:57:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08f6893a4c2c4c9f9483a6d3ade92a38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f24470f3ef94334b2e8536131d79d05 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80964459d86e496bbfb8f4d80cc9de42 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24fcdb8e25014cc79a8f619f8e617bc8 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/327d74e516e24bac8016753f5cb70590 : SUCCESS in 2m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d84fd776d6f04e9b8a042493f55971e4 : SUCCESS in 1h 22m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2c9f8ec256242ec95e6f82397c54daf : SUCCESS in 3m 53s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/50c9931726fd45beb2a0dfac42088406 : SUCCESS in 57m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f5a8f2b9d054456baa38ffaa64975ba : FAILURE in 37m 36s","accounts_in_message":[],"_revision_number":2},{"id":"8bf89681a65432eadcff6a0bae56149c96c0c4aa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-24 15:35:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b96878d27def7fe5511a36f7eed76f9f579a7be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 17:01:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91981e7160fb4dfda20f502647acdd6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b3ea61c95ad4e63bcf357560c7a24c0 : SUCCESS in 5m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a160b9c358a34ee591bd852ba3f378b8 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40c344700d6e44ee961e6b529d00ce2a : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7ef2fcddc6848d3888d8bbf0e893a4b : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9ff4dd37faf428ab3515d1f3e90265b : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/671227fb0dfd4b12a6f86531053ba3cb : SUCCESS in 3m 40s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b0e7fddabfd4ed19c7ad3a83f754635 : SUCCESS in 1h 16m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a436721b489c4a47bf3a279f7c0ec68c : SUCCESS in 3m 28s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/44cf9c581bf349b2b3fbaf5ef8cdf347 : SUCCESS in 1h 23m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c5c835c413b4d6684b34fce6468cae0 : SUCCESS in 40m 00s","accounts_in_message":[],"_revision_number":2},{"id":"10e810b338cd46e9eb12d81403e7b88c5005c361","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 17:01:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7866ccba74f325eb70408dfe246b8f0bb466d164","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 18:24:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/900db778493c422fb2806bcc1e8ac403\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e406019bfcd4b358cd18a56eeb8e3b9 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af0c90674fcb4666a1692fef1489b0cd : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa0bcbaf14914bcab058f30605f2c451 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e1e5c841b0249b685078312b987bf6c : SUCCESS in 3m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/477a0b4c5cc4488e975e2183609291f5 : SUCCESS in 1h 21m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33ffd9b16cd040939690b2debab14bb2 : SUCCESS in 3m 29s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ebea2916e6e94176ab025ca3552815db : SUCCESS in 1h 13m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/731fd0d5451b44818c89ba10dfd5172f : SUCCESS in 43m 24s","accounts_in_message":[],"_revision_number":2},{"id":"28ecc6b113aef8229fc41ca58004dcb6aca8d07f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 18:24:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f8e4b67bb21f3e2c3c1f6898ef2eca1b1449b2ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 18:25:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca4253cd757d48a094d73cbdd986c111\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ecb638a7ab7475286d8ca83fc83b402 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12f0ca77f14044aaa6bb4f3dcc1c074a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"057da4b7656110bfb125346fcbe2394312a025f9","revisions":{"3f9cbe45252d93dac4c8dbd11c1629aec43137b0":{"kind":"REWORK","_number":1,"created":"2022-11-16 14:02:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/38/864738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/864738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/864738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/864738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/864738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/864738/1"}}},"commit":{"parents":[{"commit":"f7c009ba2485f68d172924d9342d81e55a62044d","subject":"Enforce policy for qos_policy_id attribute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7c009ba2485f68d172924d9342d81e55a62044d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 10:14:02.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-13 07:59:57.000000000","tz":60},"subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","message":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"\n\nThe plugin \"qos\" is added to the \"EXT_PARENT_RESOURCE_MAPPING\" constant.\nThe policy engine can now check the QoS rule ownership using the QoS\npolicy project ID. The QoS rules are resources that do not have an\nassigned project ID, it is inherited from the QoS policy. This patch\nallows to check a QoS rule ownership using the \"ext_parent_policy_id\"\nfield, that stores the QoS policy project ID.\n\nNOTE: once released, this bug fix must include Neutron unit tests\nchecking, using the rules:\n  \"update_policy_bandwidth_limit_rule\": \"rule:admin_or_owner\"\n  \"update_policy_packet_rate_limit_rule\": \"rule:admin_or_owner\"\n  \"update_policy_dscp_marking_rule\": \"rule:admin_or_owner\"\n  \"update_policy_minimum_bandwidth_rule\": \"rule:admin_or_owner\"\n  \"update_policy_minimum_packet_rate_rule\": \"rule:admin_or_owner\"\n\nConflicts:\n       neutron_lib/services/constants.py\n\nCloses-Bug: #1996606\nChange-Id: I0531ea2c1bf29aecfab9b491eefef044a4ee49ad\n(cherry picked from commit 7e4a6beed43d28f7ab0c6c47ea056c70ea7b1fea)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3f9cbe45252d93dac4c8dbd11c1629aec43137b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3f9cbe45252d93dac4c8dbd11c1629aec43137b0"}]},"branch":"refs/heads/stable/xena"},"057da4b7656110bfb125346fcbe2394312a025f9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-17 13:22:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/38/864738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/864738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/864738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/864738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/864738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/864738/2"}}},"commit":{"parents":[{"commit":"476c87a14dd34b574968e61eac132fc671d9ea47","subject":"Fix pep8 job issues with pylint\u003d\u003d2.15.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/476c87a14dd34b574968e61eac132fc671d9ea47"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 10:14:02.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-11-17 13:22:52.000000000","tz":0},"subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","message":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"\n\nThe plugin \"qos\" is added to the \"EXT_PARENT_RESOURCE_MAPPING\" constant.\nThe policy engine can now check the QoS rule ownership using the QoS\npolicy project ID. The QoS rules are resources that do not have an\nassigned project ID, it is inherited from the QoS policy. This patch\nallows to check a QoS rule ownership using the \"ext_parent_policy_id\"\nfield, that stores the QoS policy project ID.\n\nNOTE: once released, this bug fix must include Neutron unit tests\nchecking, using the rules:\n  \"update_policy_bandwidth_limit_rule\": \"rule:admin_or_owner\"\n  \"update_policy_packet_rate_limit_rule\": \"rule:admin_or_owner\"\n  \"update_policy_dscp_marking_rule\": \"rule:admin_or_owner\"\n  \"update_policy_minimum_bandwidth_rule\": \"rule:admin_or_owner\"\n  \"update_policy_minimum_packet_rate_rule\": \"rule:admin_or_owner\"\n\nConflicts:\n       neutron_lib/services/constants.py\n\nCloses-Bug: #1996606\nChange-Id: I0531ea2c1bf29aecfab9b491eefef044a4ee49ad\n(cherry picked from commit 7e4a6beed43d28f7ab0c6c47ea056c70ea7b1fea)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/057da4b7656110bfb125346fcbe2394312a025f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/057da4b7656110bfb125346fcbe2394312a025f9"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
