)]}'
{"id":"openstack%2Fneutron-lib~864739","triplet_id":"openstack%2Fneutron-lib~stable%2Fwallaby~I0531ea2c1bf29aecfab9b491eefef044a4ee49ad","project":"openstack/neutron-lib","branch":"stable/wallaby","topic":"bug/1996606","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-11-23 20:51:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0531ea2c1bf29aecfab9b491eefef044a4ee49ad","subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","status":"MERGED","created":"2022-11-16 14:03:24.000000000","updated":"2022-11-23 20:53:00.000000000","submitted":"2022-11-23 20:51:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864739-bug/1996606","meta_rev_id":"6aaa2e899bc4127f9882ba27c0fde3ede428a1cb","_number":864739,"virtual_id_number":864739,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-23 20:51:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-11-22 14:13:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-11-23 18:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-11-23 18:32:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-16 15:27:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-22 14:13:11.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-11-23 18:32:07.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"c3a7f13d7516849cb9969c78c61f4789024216a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-16 14:03:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"722c271dcd9a205199510111240ef941aae4777f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 15:27:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a2ea6b69a9a42fe9f6c43f0e3e7732c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/287cad227daa48709d0d6300f9d01ebc : SUCCESS in 4m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07f6e288a6d94e1d89a1d5860ed00da2 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15360295813a496ab9e8557b4efc4e31 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56b0b70e59784eb990fe0f69fe1d6f09 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d972655a5fbf4139b43efc4aa06e049d : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb763e8b8acd404eb6662a51312b57eb : SUCCESS in 6m 36s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c08fd822ec64afb8305be9fb5f017c6 : SUCCESS in 1h 20m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84ca434114974f859a4ded7e15d129a8 : SUCCESS in 2m 49s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/9fc730e73273478aabc71422cc67b4fc : SUCCESS in 47m 13s","accounts_in_message":[],"_revision_number":1},{"id":"60e362e028c65ade3b59efd1676b475a4d944e7c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-22 14:13:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c449b673e98a0672cbd986b8142687d4dd7e542b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-11-23 18:32:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c37ff93363b9e9907148b1fdbc06cfc180989797","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 18:32:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8e838481c75d3c5e0787a91cce9149ef884602f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 20:51:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcb855e68365459e90b74a1c432b85f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/343dee1f2a0b4e188c76bc4de751bc12 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81ec0b30b57a4f279c738178daa80c05 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6e22f123adc454da0afb7b35a396063 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bce294df2b74624962456ca9e868d36 : SUCCESS in 3m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9619a72cbd214a33975441a12fc59cc1 : SUCCESS in 1h 28m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6a99c79ebf0491aac8b91f0b8e121b0 : SUCCESS in 4m 58s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/a0ab401fad1a49528c5f3eda69bf4595 : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c01eb71f33bc835a4840c6ca33a2c8a571d01ce4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 20:51:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6aaa2e899bc4127f9882ba27c0fde3ede428a1cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 20:53:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8890aca5059542dea283434bc0769e67\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a171e3ded60349678e24841a1fed9b64 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/959dd160efcd4889812b9716a52b78ca : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"acfb2fdc29ab6b22003903b16688cde373f5082a","revisions":{"acfb2fdc29ab6b22003903b16688cde373f5082a":{"kind":"REWORK","_number":1,"created":"2022-11-16 14:03:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/864739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/39/864739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/39/864739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/39/864739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/39/864739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/39/864739/1"}}},"commit":{"parents":[{"commit":"6bbae46890f458cbb0d7cb5dd87e80f07a305b53","subject":"Enforce policy for qos_policy_id attribute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6bbae46890f458cbb0d7cb5dd87e80f07a305b53"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 10:14:02.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-13 08:00:50.000000000","tz":60},"subject":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"","message":"Added \"qos\" plugin to \"EXT_PARENT_RESOURCE_MAPPING\"\n\nThe plugin \"qos\" is added to the \"EXT_PARENT_RESOURCE_MAPPING\" constant.\nThe policy engine can now check the QoS rule ownership using the QoS\npolicy project ID. The QoS rules are resources that do not have an\nassigned project ID, it is inherited from the QoS policy. This patch\nallows to check a QoS rule ownership using the \"ext_parent_policy_id\"\nfield, that stores the QoS policy project ID.\n\nNOTE: once released, this bug fix must include Neutron unit tests\nchecking, using the rules:\n  \"update_policy_bandwidth_limit_rule\": \"rule:admin_or_owner\"\n  \"update_policy_packet_rate_limit_rule\": \"rule:admin_or_owner\"\n  \"update_policy_dscp_marking_rule\": \"rule:admin_or_owner\"\n  \"update_policy_minimum_bandwidth_rule\": \"rule:admin_or_owner\"\n  \"update_policy_minimum_packet_rate_rule\": \"rule:admin_or_owner\"\n\nConflicts:\n       neutron_lib/services/constants.py\n\nCloses-Bug: #1996606\nChange-Id: I0531ea2c1bf29aecfab9b491eefef044a4ee49ad\n(cherry picked from commit 7e4a6beed43d28f7ab0c6c47ea056c70ea7b1fea)\n(cherry picked from commit 3f9cbe45252d93dac4c8dbd11c1629aec43137b0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/acfb2fdc29ab6b22003903b16688cde373f5082a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/acfb2fdc29ab6b22003903b16688cde373f5082a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
