)]}'
{"id":"openstack%2Fneutron-lib~864764","triplet_id":"openstack%2Fneutron-lib~stable%2Fxena~I3a4a27ec4672f8ea8848d7c04651730dae6f40ff","project":"openstack/neutron-lib","branch":"stable/xena","topic":"bug/1988296","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-11-17 12:46:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a4a27ec4672f8ea8848d7c04651730dae6f40ff","subject":"Fix pep8 job issues with pylint\u003d\u003d2.15.0","status":"MERGED","created":"2022-11-16 16:42:25.000000000","updated":"2022-11-17 12:47:56.000000000","submitted":"2022-11-17 12:46:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864764-bug/1988296","meta_rev_id":"553794e9752d37f539b81ef7ca9c3cac3b3a0609","_number":864764,"virtual_id_number":864764,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-17 12:46:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-11-17 11:09:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-11-17 10:24:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-11-17 11:09:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-16 18:06:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-17 10:24:44.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-11-17 11:09:13.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"3ca8690d7040065088e4c1c34c01ca7fce95d3ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-16 16:42:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b13d7e6f9c1bd3cc2d71efb024fdbec632b6c95c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 18:06:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/494960a91cdb4f0d82f44618385cdc25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f35fc01d5b64385ad26ba20a2193a7a : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0309f9c5f9ed49b9b2f63e6eadcb926b : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eb27a2c76cf4e39b9a9f1de6aa3a59a : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8570d34521a49c4bee8396389f436d3 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/532d4599eb1f45d19dcf9e789f215cd8 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b2422916d7e4c31b20c300d34e0c7ea : SUCCESS in 3m 41s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96f10b30cb0941679a937269c37fca1e : SUCCESS in 1h 21m 03s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a10fe24035b74581b7c78b09823cc942 : SUCCESS in 1h 12m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac32e9f1bafe4c7aba58807b192eb32f : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9e9dbb8abd1d11fe01b2016eb2b3ad3e6af1a111","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-17 10:24:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"805e27559aeadff9b9e7dd495a2755c23d6ef043","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-17 11:09:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7d4c2567938650af0709603fec206077cc60af2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 11:09:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9bd2830b57625549bef7408cd051cc8e78cb32f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 12:46:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f53d40b0f8a94e58a11e3169e2f7080e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a64824ea090482f85e9e60441204637 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83656b5502324fe5bf046d95e0edb3b2 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfe436c6ec5a4ea9ab9f44e0f302581c : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29979201396e4c378c4e3a84cfc24be2 : SUCCESS in 3m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27ac903bb7aa4b0ba8371e3dee1adab0 : SUCCESS in 1h 24m 27s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/30c604e33bcf4542a2b63505a9559bf2 : SUCCESS in 1h 15m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f37d0b44e8e7423c89f41939c7e72aa4 : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":1},{"id":"65a70de199c3eb25101df8f2161076ad4045b832","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 12:46:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"553794e9752d37f539b81ef7ca9c3cac3b3a0609","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 12:47:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9be4e1e17a8248a7b00a8b7b618ca966\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f01def419884d248853798371885c16 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"476c87a14dd34b574968e61eac132fc671d9ea47","revisions":{"476c87a14dd34b574968e61eac132fc671d9ea47":{"kind":"REWORK","_number":1,"created":"2022-11-16 16:42:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/64/864764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/64/864764/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/64/864764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/64/864764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/64/864764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/64/864764/1"}}},"commit":{"parents":[{"commit":"f7c009ba2485f68d172924d9342d81e55a62044d","subject":"Enforce policy for qos_policy_id attribute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7c009ba2485f68d172924d9342d81e55a62044d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-31 10:57:04.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-13 10:39:44.000000000","tz":60},"subject":"Fix pep8 job issues with pylint\u003d\u003d2.15.0","message":"Fix pep8 job issues with pylint\u003d\u003d2.15.0\n\nThis patch also includes [0], needed to unblock the CI for\nstable/yoga. \"oslo.context\" has removed the deprecated argument\n\"tenant\" from RequestContext[1]. This patch is included in\n\"oslo.context\" 4.1.0 [2].\n\nThis patch also includes [3], that fixes the \"redundant-keyword-arg\"\npylint error in the \"update_attributes_map\" method.\n\n[0]https://review.opendev.org/c/openstack/neutron-lib/+/864751\n[1]https://review.opendev.org/c/openstack/oslo.context/+/815938\n[2]https://review.opendev.org/c/openstack/requirements/+/829599\n[3]https://review.opendev.org/c/openstack/neutron-lib/+/840816\n\nCloses-Bug: #1988296\nCloses-Bug: #1996776\n\nChange-Id: I3a4a27ec4672f8ea8848d7c04651730dae6f40ff\n(cherry picked from commit 7026199065fc67977871c6f2521ce6334c20147a)\n(cherry picked from commit a8abe9d592da5bcf065af40e0ba1cd3599ede1e7)\n(cherry picked from commit ef2d826a453cde02077c2b12d0508bd235f047ba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/476c87a14dd34b574968e61eac132fc671d9ea47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/476c87a14dd34b574968e61eac132fc671d9ea47"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
