)]}'
{"id":"openstack%2Fneutron-lib~869294","triplet_id":"openstack%2Fneutron-lib~master~I9be781d99a08a6a015c3747eb15f20bc356cf08e","project":"openstack/neutron-lib","branch":"master","topic":"manila-multiple-port-binding","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2023-01-24 15:01:29.000000000","reason":"\u003cGERRIT_ACCOUNT_18816\u003e replied on the change","reason_account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"}}},"hashtags":[],"change_id":"I9be781d99a08a6a015c3747eb15f20bc356cf08e","subject":"add DEVICE_OWNER_MANILA_PREFIX to constants","status":"MERGED","created":"2023-01-06 13:41:08.000000000","updated":"2023-01-24 18:39:36.000000000","submitted":"2023-01-24 18:38:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869294-manila-multiple-port-binding","meta_rev_id":"2a6e192ee910e72b750fbccdb64ff040bcf78462","_number":869294,"virtual_id_number":869294,"owner":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-24 18:38:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2023-01-24 12:05:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-01-16 16:21:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":2,"date":"2023-01-24 12:33:23.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":1,"date":"2023-01-24 12:33:23.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-06 15:20:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-06 15:51:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-01-12 15:30:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-01-16 16:21:17.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2023-01-24 12:05:29.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-01-24 12:33:23.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"0bd06a54abd12adee28a9dc7f2bf7a87549e292c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-06 13:41:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cde702ce72071bf2330e5af999098e50826b18ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 15:20:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39c313931dac4641b4fa75a811681c10\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d613df77aef4d9daa01951439d10022 : FAILURE in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86fb39dd017e4e66a302c657eefb6767 : FAILURE in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e8ec661155421abf40f65de33a93c6 : FAILURE in 1m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ce8715d99364340a2e6c63335c352e4 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4caf6ee7fad64292b3446a7dac9c85cd : FAILURE in 2m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2a594486858436e967719835e9df26c : SUCCESS in 1h 37m 08s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/7920dde876bc47909d9e42fee2c68463 : SUCCESS in 1h 15m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/91ddbe5de0554efb981829f50e490137 : SUCCESS in 1h 02m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/32d00184d32f47e695271f44efe0f5c9 : SUCCESS in 57m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6ae3cc440c69434ba6c4463c5b354161 : SUCCESS in 1h 02m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/877ae05e49b7498e89397d6afa0db087 : SUCCESS in 39m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/01ce5da3ed9d4dc6806a6f9cf8889390 : SUCCESS in 57m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5dab39adb70105a3eecb3f489d54ae6a7e62d10f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-01-06 15:51:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e920a10676343b66ce35ffad645b3839027b9ac2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-12 15:30:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"238398d23949579a4c80e8feadecd9cabfed2e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 17:11:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa807529827e4007bccd59018552f046\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9dcfcd1efb6421db104178fbb6566ad : SUCCESS in 6m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/937f14ac93c443ff8f99c6ef65878013 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5be78eed244943dfa9f3211891b5503f : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c54ac2bc797c450fac83adde8eb5a1fe : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0969326e015b481fb54b8663e077ce97 : SUCCESS in 3m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b332a14a5a52408ab8a175b88ed38343 : SUCCESS in 1h 36m 21s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c279c1353f2b429e85b3fffd2ced11c0 : SUCCESS in 1h 02m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/731760c7be8c42209477e5fb46b01913 : SUCCESS in 1h 06m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dbd51dbdcefb44aebfd7ef366aea173f : SUCCESS in 1h 07m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cf1046ea2ce1428cb13e03b1fef054be : SUCCESS in 1h 00m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/71d48be6726e41f285811fd3ee41b979 : SUCCESS in 26m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0fb94615c8f94981906124337778bf39 : SUCCESS in 49m 00s","accounts_in_message":[],"_revision_number":1},{"id":"df3b6c71ce32855c61bb3b438424826120bd7cb7","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-01-16 16:21:17.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ff320087d0d1bda9d322372d12cb052848a1baf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-18 06:55:49.000000000","message":"Topic bug/1580880 removed","accounts_in_message":[],"_revision_number":1},{"id":"c6149d68e4af6f6611d0e61f839ab8304d205a8b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-18 06:55:53.000000000","message":"Topic set to manila-multiple-port-binding","accounts_in_message":[],"_revision_number":1},{"id":"84e142baf7db720d8c1c8d74a81d0504c30e8bba","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-01-24 12:05:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"07557e4179625ac8bdde9ce24557b8e04642a533","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-01-24 12:33:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e629c45b6e33d28c181450c832a14d40833de32e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 12:33:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52bf8c7cdf1169b55b4739456ee74479c327ea2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 14:17:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d5c09fb4d0b4bbe9a3af6ee8a245142\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e315d73fd08a4ba69472731882c5b028 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d5fc6326355417781088d814fceafc0 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5eef4491e1c437882c3b3a2750bcc21 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/163d3c75d7f04911b0df8e66d484d2e2 : SUCCESS in 3m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f3952f7cb29494b91af28a1b9311c3e : FAILURE in 1h 14m 07s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/8e9db34c8d514a8082d4cedf4bf70088 : SUCCESS in 1h 40m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3133f9ecce7247349328e18819a01c87 : SUCCESS in 1h 29m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ebed1f022bdb4df9b3b8d0a9072da5ce : SUCCESS in 1h 23m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cd38c8c73f394e1c864aa7f08c80dddd : SUCCESS in 1h 24m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/37e5f741bb2740028df79e87eb77bf4e : SUCCESS in 56m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9d9082fdb8b9e417286fb0945776a9d91d6ab9a6","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-24 15:01:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6ca224e6b0d3092d68370241f67b778ed1d0278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 16:53:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/292350a231db41c4b23081ba5c9066f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d1aa26964c04acebc9ededdcba777d7 : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d812dbce4f144c6199be60cc6dea5699 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6b245e47e9e499d87501c206df1b7a2 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/676d7c0d13854bc5b775a7a1e2ff6e43 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2103b918cf6a4a8fbd49ef8f5193ed35 : SUCCESS in 3m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ac141e1b88b43f69ddccc6624c74f5b : SUCCESS in 1h 46m 29s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/26775c75005647ef8d99d296cdb0547b : SUCCESS in 1h 08m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f92f029b4db34f608e03231fab5713a7 : SUCCESS in 1h 40m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7b4e9f4b5e964007bc96c6af910dea22 : SUCCESS in 1h 09m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/735434d6f6ec43109318e44e8c2a11a1 : SUCCESS in 1h 28m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/86a55a9614e2498d9ae404763fbee8fd : SUCCESS in 26m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d116738191b1499d845820ce879d1623 : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":1},{"id":"491ef0ce680f9fefdbe96a6184e62c56f442922e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 16:54:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"041660db120c8517186cb8877171046d44189f21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 18:38:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e288ed99cfe1468d9447c6b8fc309bb4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a628d7f8ec465a81a9c6bdd47b5d18 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00c4953120414b6faa4a2d825995d505 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba36c4017b9f4a0f9c55558a7b660232 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c90d2b0a5f394a28932023a058193bf0 : SUCCESS in 3m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10058900c7e4414b9b6b0fa42d0ede6b : SUCCESS in 1h 42m 38s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e832487bf8bf49e7a9a54114e638b390 : SUCCESS in 1h 32m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1fe4394c4a2c483d92257666a7cd4a18 : SUCCESS in 1h 25m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/04061a8e86874b83a44d2ed8c354ce0e : SUCCESS in 1h 32m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/99d6b71b275a45eca719bc227c758d43 : SUCCESS in 1h 22m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2133f2c836c24f1482a57afbb75afb44 : SUCCESS in 42m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ead5f31e64ad961f831ecb958701e53c78a78946","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 18:38:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2a6e192ee910e72b750fbccdb64ff040bcf78462","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 18:39:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01bf7c95ee004a4ea8a1bea5905674ca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d91a8444f9cc44878820c9d872e9de10 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d58d4ac10ad8a1570e2f14d392342712bc9fe41","revisions":{"9d58d4ac10ad8a1570e2f14d392342712bc9fe41":{"kind":"REWORK","_number":1,"created":"2023-01-06 13:41:08.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/94/869294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/94/869294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/94/869294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/94/869294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/94/869294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/94/869294/1"}}},"commit":{"parents":[{"commit":"7e9f66aca498706747b23e01d7e0622c7c187002","subject":"Fix a few broken links in api-ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e9f66aca498706747b23e01d7e0622c7c187002"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2022-10-19 11:28:46.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2023-01-06 13:41:06.000000000","tz":60},"subject":"add DEVICE_OWNER_MANILA_PREFIX to constants","message":"add DEVICE_OWNER_MANILA_PREFIX to constants\n\nmanila port device owners are prefixed \u0027manila:\u0027 according to the manila\nneutron driver for binding, see\nhttps://review.opendev.org/c/openstack/manila/+/283494/29/manila/network/neutron/neutron_network_plugin.py#141\n\nPartial-Bug: #1580880\nChange-Id: I9be781d99a08a6a015c3747eb15f20bc356cf08e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9d58d4ac10ad8a1570e2f14d392342712bc9fe41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9d58d4ac10ad8a1570e2f14d392342712bc9fe41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
