)]}'
{"id":"openstack%2Fneutron-lib~869481","triplet_id":"openstack%2Fneutron-lib~master~I18765db9682fced05e8441a1e2122f9f814cd31b","project":"openstack/neutron-lib","branch":"master","topic":"tox4","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-01-16 11:35:13.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-01-16 12:54:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I18765db9682fced05e8441a1e2122f9f814cd31b","subject":"Update envlist for tox4","status":"MERGED","created":"2023-01-06 15:41:29.000000000","updated":"2023-01-16 12:55:52.000000000","submitted":"2023-01-16 12:54:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"869481-tox4","meta_rev_id":"934bbcad3d45077a93318c11a2b575c8e536094e","_number":869481,"virtual_id_number":869481,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 12:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2023-01-16 11:35:13.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-01-16 10:54:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-01-16 08:40:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-01-16 10:54:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-06 16:06:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-01-06 17:28:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-09 07:14:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-01-12 16:03:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-01-16 10:54:58.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"c505bc6b8271a7fd471badf7e4508b86f5879f6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-01-06 15:41:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb6682881a63bebc0b5bfbbbfd5a6fee2dac0d0a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-01-06 16:06:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f6292dd28d1ec57381a8c1c33ee3d14cb5286ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 17:28:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c74132be0744a66acf1b9ecc0f9cc0d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/49bbb310cead4116950408d6b6329740 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53675b99069f4652aa51cb27adfef3ad : SUCCESS in 6m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a81ad33b25147cf998fd84488442b0f : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61bd8a9acfe14035af56c77f7dc16aec : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f40a2873a59c439c8d0af5e09e025b9b : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69e44525563c486da205cc880857ccc4 : SUCCESS in 3m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/639d3609306b48c1a0548460c9819b0d : SUCCESS in 1h 41m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57d7bd5b129a4a81961e97d17844c8e0 : SUCCESS in 2m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d098f34b358a432c9203852dc310a7ef : SUCCESS in 54m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7434a1d9572d651f1c12164dd7921930a6fcd210","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-09 07:14:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b30922abccbc25032a5a155b810738ef37c8723e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-01-10 09:14:44.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"153bc62d3967f89505d7d0d12d3cf2928f510dfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-01-12 15:15:32.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"090a7c6787c2510586fae8ee0402c5c3b467b673","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-12 15:56:34.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cde0e0d7340cf3d8cea7a4dbde89ef7e2fd0f49c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-12 16:03:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c2a4eab3206796fd34c0955b678993d52d5aad52","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-01-12 16:29:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"461e07c13936e8d0da11d9e1b140ef0a583e19a7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-01-12 16:44:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5db63b70bb788f3b53baa6e308526aaf3692c9ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 16:55:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19f64aa21bf8447db1e156788351e0cd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f814374d2631480e94a416e46e938d1e : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f6c634ff6324c87b65286d76ca9c48e : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59c92d33099e492a864eb729bfae0667 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f930b72994a2406fa3be9acb82e3a0bc : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/450ba57051b74ee6ad2095f3adb98b04 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c70ff64470b4b0eb4f6a82d1020b45f : SUCCESS in 3m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/825d6fff46d540fc8dc3a7b6094d2844 : SUCCESS in 1h 34m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba22250198624cfc80dbcf8ae3d9a5f8 : SUCCESS in 2m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/369ea553844e4563ada89f86d92b457d : SUCCESS in 48m 43s","accounts_in_message":[],"_revision_number":2},{"id":"d8c2a8b6246608b9985dab4f83ffc0ddfbf10897","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-01-13 14:58:14.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ec88bf6cac86dc62d058ecbb7e21a4b0236243c3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-01-13 14:58:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"57138576c4775a1ceedf7ff608b3f96216832cc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 17:11:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0047626ff63041569a056951e4a7da7e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d4f97235ce4b45acbcdb1a8abd5b6a2e : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee828e3f6f154ce8b78d3dcad38c07d0 : SUCCESS in 5m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c3f610e2f9e4119854fa70461482f28 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89d610b114954d27823039d0aa9dfff7 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/087db24541224431a14ebc9ad38ca90a : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8853d13c87954e9281791ee085baeccf : SUCCESS in 3m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/774337ee5c324a88b3eb045e4a15d340 : SUCCESS in 1h 56m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/113b27b137894d8cb6b55bfef22aabf8 : SUCCESS in 4m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b1ed665a17e46d789647141c3af8dc4 : SUCCESS in 48m 00s","accounts_in_message":[],"_revision_number":3},{"id":"9192c686dd03038b0ef49f6836b67538c59bb320","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-01-16 08:40:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8a243d882b8d03f9a771c5f6577caa0a254d4b18","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-01-16 10:54:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"85335603aa2376353a73fbbde9dd334340532a42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 10:55:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f23b404647e0403501ddff80d034aac941bce5d3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-16 11:35:13.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c68a02e3c638d34368ea5dd5ba5dbdf24ec02924","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:54:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e795ed90ae6f4203b9385e7552e512ac\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7d70c95a50024142bfe55802cd30690f : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7428e3e9511c4bb39ba397d364ddcc45 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1de855dbc151461eb9362d8aca290e12 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba5e40711a224c49a26906676f34c018 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f7a0cf03497438fb69ced85606aef45 : SUCCESS in 4m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a12a920bdc44a29864e864d19db973d : SUCCESS in 1h 53m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bc5ac0c9ec242409da56417de39f70c : SUCCESS in 2m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c482e2ee66e349aaa04923e2b95bcaf7 : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":3},{"id":"ce2a4a2d049f8da4db79c87c5c5436ecc259ff84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:54:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"934bbcad3d45077a93318c11a2b575c8e536094e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:55:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4c43cd4c8e9407baf06a696fe052804\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6caac6d2c36c491a83b8ee2c7470e3ae : SUCCESS in 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ceb18269b8346c4aa9d50138f562452 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3180eb5147034f64ba61f190c5b9d324 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4ac49dbcc08dcb2556806397881faaaa456cc724","revisions":{"8832972c9ff3d166d488b82347adca42788661bb":{"kind":"REWORK","_number":1,"created":"2023-01-06 15:41:29.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/81/869481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/81/869481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/869481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/869481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/869481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/81/869481/1"}}},"commit":{"parents":[{"commit":"7e9f66aca498706747b23e01d7e0622c7c187002","subject":"Fix a few broken links in api-ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e9f66aca498706747b23e01d7e0622c7c187002"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-01-06 15:40:42.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-01-06 15:40:42.000000000","tz":-300},"subject":"Temporary update envlist for tox4","message":"Temporary update envlist for tox4\n\ntox4 even when running specific env with -e option tries\nto find python version for all envs defined in envlist[1]\nwhen running with --skip-missing-interpreters\u003dfalse and\nfails if interpreter missing for any of the env.\n\nWith py38 in envlist it tries to find python3.8\nwhich is not available in ubuntu-jammy by default\nand fails, hence use generic version instead i.e py3\nto handle it.\n\nOnce [1] is fixed we can update the list as per\ndefault supported py3 version for a release.\n\n[1] https://github.com/tox-dev/tox/issues/2811\n\nChange-Id: I18765db9682fced05e8441a1e2122f9f814cd31b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8832972c9ff3d166d488b82347adca42788661bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8832972c9ff3d166d488b82347adca42788661bb"}]},"branch":"refs/heads/master"},"73037acc398f073a9c8d1fa32925a21cd37fe989":{"kind":"REWORK","_number":2,"created":"2023-01-12 15:15:32.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/81/869481/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/81/869481/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/869481/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/869481/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/869481/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/81/869481/2"}}},"commit":{"parents":[{"commit":"7e9f66aca498706747b23e01d7e0622c7c187002","subject":"Fix a few broken links in api-ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e9f66aca498706747b23e01d7e0622c7c187002"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-01-06 15:40:42.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-01-12 15:15:24.000000000","tz":0},"subject":"Temporary update envlist for tox4","message":"Temporary update envlist for tox4\n\ntox4 even when running specific env with -e option tries\nto find python version for all envs defined in envlist[1]\nwhen running with --skip-missing-interpreters\u003dfalse and\nfails if interpreter missing for any of the env.\n\nWith py38 in envlist it tries to find python3.8\nwhich is not available in ubuntu-jammy by default\nand fails, hence use generic version instead i.e py3\nto handle it.\n\nOnce [1] is fixed we can update the list as per\ndefault supported py3 version for a release.\n\n[1] https://github.com/tox-dev/tox/issues/2811\n\nChange-Id: I18765db9682fced05e8441a1e2122f9f814cd31b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/73037acc398f073a9c8d1fa32925a21cd37fe989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/73037acc398f073a9c8d1fa32925a21cd37fe989"}]},"branch":"refs/heads/master"},"4ac49dbcc08dcb2556806397881faaaa456cc724":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-01-13 14:58:14.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/81/869481/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/81/869481/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/869481/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/869481/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/869481/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/81/869481/3"}}},"commit":{"parents":[{"commit":"7e9f66aca498706747b23e01d7e0622c7c187002","subject":"Fix a few broken links in api-ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e9f66aca498706747b23e01d7e0622c7c187002"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-01-06 15:40:42.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-01-13 14:56:43.000000000","tz":-300},"subject":"Update envlist for tox4","message":"Update envlist for tox4\n\ntox4 even when running specific env with -e option tries\nto find python version for all envs defined in envlist[1]\nwhen running with --skip-missing-interpreters\u003dfalse and\nfails if interpreter missing for any of the env.\n\nWith py38 in envlist it tries to find python3.8\nwhich is not available in ubuntu-jammy by default\nand fails, hence use generic version instead i.e py3\nto handle it.\n\nEven though [1] is fixed we are still seeing this failure.\n\n[1] https://github.com/tox-dev/tox/issues/2811\n\nAlso updated basepython to use TOX_PYTHON if set.\n\nChange-Id: I18765db9682fced05e8441a1e2122f9f814cd31b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4ac49dbcc08dcb2556806397881faaaa456cc724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4ac49dbcc08dcb2556806397881faaaa456cc724"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
