)]}'
{"id":"openstack%2Fneutron-lib~870080","triplet_id":"openstack%2Fneutron-lib~master~I5cb21b25fcda62eae3743b2f96177dcc1fda4afb","project":"openstack/neutron-lib","branch":"master","topic":"port-hints","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2023-02-03 11:52:35.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-02-03 11:52:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5cb21b25fcda62eae3743b2f96177dcc1fda4afb","subject":"port-hints: api definition","status":"MERGED","created":"2023-01-13 11:52:35.000000000","updated":"2023-02-03 11:53:49.000000000","submitted":"2023-02-03 11:52:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"870080-port-hints","meta_rev_id":"2ced2460ab672032b2011fa16144ccd63e5aa532","_number":870080,"virtual_id_number":870080,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-03 11:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-02-03 09:25:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-02-02 10:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-02-02 10:24:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-02 10:24:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-02-03 09:25:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-13 14:06:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-24 12:47:16.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-01-31 01:15:53.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-01-31 02:25:26.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-02-01 08:02:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"7a79cc8f2798c4cccdc4520bd43c21c43f82f88e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-01-13 11:52:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf2ed490bba846208510179baf9f569de3bdb7ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 14:06:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a55b940db2384463acf2a5724e5435b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/621342e55aef46a68213e2ab1581cea7 : SUCCESS in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73f58e184549405997bbd65051c6e19c : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ce3dda6358b4655936bc85f635e4004 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8871a7b353d465da3b40e2ec08cdf91 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f9e614cefb44ed3a34cf9721d9f5a1b : SUCCESS in 3m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfc8a0030b1f4ddbaf53f68c71d37d17 : SUCCESS in 1h 54m 16s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b083cd63fffe4a7a9da49cafb917282f : SUCCESS in 1h 35m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fc6e9e9bc84f4b769a2d090637f4ab10 : SUCCESS in 2h 09m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f128833d55f64049ba272a4ed1cfb754 : SUCCESS in 1h 40m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a5720348a1584499b091e1f3a2e5c093 : SUCCESS in 2h 09m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bbc96b94cafa422d81896501f705b85f : SUCCESS in 47m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/abd32c155d304abfa39ed38dfe25c413 : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":1},{"id":"1c8af4f622bf2655654ee232244b52f2086be63f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-01-18 13:40:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3a6de6249bd2cedfb994668a3fb77cd45ae4aaa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 15:25:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa1e303e923f48dfb0926bc951daa77f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a1957ad903094720a09e22616966aede : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8474aa9d598b4a6b87026bfab024fc76 : SUCCESS in 5m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae26b2f90ec041d6a98897300f4ba94d : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aaccc37b2a34ef9bed75c0b38c6cffe : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48cdf3dc32954371918b0bf9de78812d : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d4918a3e4664ca6ac772ff8a53948ba : SUCCESS in 3m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50e638ba667e4ca9b3826df14f99bebc : SUCCESS in 1h 42m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b9d6e67e2704497896893afe43bf3cd : SUCCESS in 4m 07s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b0f0f499c3af4b7087977cc440fec5f2 : SUCCESS in 1h 15m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9adfe8132e87475d966569065c6e0fcc : SUCCESS in 1h 22m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/11f5ed660aff478f95ad43ef53090a97 : SUCCESS in 1h 35m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f8dae85138e74a42b355ba638d0fd972 : SUCCESS in 1h 06m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57fde1897b2d44979dccf7bd36f08e02 : SUCCESS in 28m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2098aba1b6fb496cb647b4689fca722e : SUCCESS in 45m 08s","accounts_in_message":[],"_revision_number":2},{"id":"95e2e767c78ee83aea5f6f4c6ea36b94cfa92274","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-01-24 12:47:16.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cecdb0b175cce398e2bb2345d6d46791b64f0025","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-01-26 10:00:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a686ba2e411ef8ec2150dd3f2a48e03bdeee0d2a","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-01-31 01:15:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"678accffef81019fe71147a51e6337eb493a6512","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-01-31 02:25:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"35d4ca9ae16337139f6bb5e9ca3d1db228a669f9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-01 08:02:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f81a85fe25cbbcd6d267c4cfb1f17518f7837bb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 08:03:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cfc2fc8c998c9ca2faab5e5f8f6caeb3042369bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 10:13:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4b804a4dc064b76bca4e55f45fc2291\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2c5388fc8317425d9911f9f0be280874 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/841430af604d486aafcda1174707745b : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7c8b2a147654793a94b1da67c496e89 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08f523d3463e4c97a415975de4b6822e : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/290b9bf5a8574aa09537d12bc22b729c : SUCCESS in 4m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/224a921081464c65a24f4ddf7c2ba45c : SUCCESS in 1h 39m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f94d478d65d41f7bcbb46901f4642d9 : SUCCESS in 3m 29s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/dc149084d1ad40069c94eeb86e573b45 : SUCCESS in 1h 26m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4749153bd0f6493da671c77752971a60 : SUCCESS in 1h 42m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a8b727bf2911429a98237adb4de56495 : SUCCESS in 1h 14m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/25f94bd8a6dd4ed7807e37dbc784e976 : SUCCESS in 2h 08m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fe00ee353dcb47669972d6306c02dede : FAILURE in 56m 24s","accounts_in_message":[],"_revision_number":2},{"id":"217065a6bec350cb89c8dca7628ed790be14fdf2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-01 11:35:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"734445e23b29847ffd94dc543c92d2beb80849af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 12:57:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b57f15e22024fb3be9b2f46e85d7e38\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7427bebb61284090b2aca5a974c7e918 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdfa704b2c70462a8e8f173495cdce7f : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6988e12ef7ab4c6799c76419ce269b38 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d1a329d28a74ccab94b966749132aff : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3076b50b6e140e4b5d6060f40ab692a : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f85a07da0e1b46beb18e48e4e5bf3df2 : SUCCESS in 3m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/448c287b71b3499ea5f1fe8673488c80 : SUCCESS in 1h 18m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52d482ebbae347fc8fb070376d65eec4 : SUCCESS in 3m 01s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a918af42344a460eb524b6f4bc4e33d6 : SUCCESS in 1h 11m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f4edd0c0a1234c43934b6c97bf26c5ee : SUCCESS in 1h 12m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/17f3a240c90245babb7588ef8bd307e6 : SUCCESS in 1h 04m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7c6d2f34f550451d847dc2de148415c7 : SUCCESS in 1h 08m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d9f12d9311c543c3b8b4656672613599 : SUCCESS in 27m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a878c7897d6b4bd7b4206a74a7ba1d9c : SUCCESS in 39m 27s","accounts_in_message":[],"_revision_number":2},{"id":"8c156b9b4e9eab89002259a963680517873540fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 12:57:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"411ebe828a817d9ef058fff4eedb79ede7ab4327","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 15:01:58.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6c67595928647f29b52a49f3d39e99e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3df2594621cc438ca8124eb0693a4300 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e20fa1f8ee5449e3a9b98711f2980073 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f589dd2be124568be021f33dd124b14 : FAILURE in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd0ca6f3044a475cb60b9b53677101c4 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55c2ad6a7d2a4ce2a3ba494c101892c7 : SUCCESS in 3m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4b32a1e603c4886b2492368d0094195 : SUCCESS in 1h 29m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04a6e843a3174bd69d0002f272242be6 : SUCCESS in 3m 03s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9fd41a06a7084084963a627b980ffec6 : SUCCESS in 2h 01m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5761b3f427ff45c588763002280ea736 : SUCCESS in 1h 41m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a37e16375f6546419e59424281c3e199 : SUCCESS in 1h 44m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a5439974422f42f588d262f72d1e5685 : SUCCESS in 1h 29m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff3125d637a54ebd9864f14d3468be28 : SUCCESS in 1h 13m 34s\n\nWarning:\n  Comments left for invalid file neutron_lib/api/attributes.py\n  Comments left for invalid file neutron_lib/db/standard_attr.py","accounts_in_message":[],"_revision_number":2},{"id":"7c35f104ebf0142f9da01784849ea2be952b8df8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-02 09:06:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0f668be1ec9f51070b04967c64bf606edf220e3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-02-02 10:24:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"4136f7f44b679ddcfc3cd49d63901dc60d0b8887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 12:15:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0659f934fdf4bf1adfd32a331dde0e4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/788e4d85072c4c9fb9667531476f2a93 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53f61bbf688432caf60fb5f9d5bcfbb : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1c370d0eef44beea57fca9d27b9c527 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0bef66c0f844aedb89c5f6d1d29e0b2 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/658346e6c8434a08b27f45819d0922a1 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24c424ef23d947f594dbfa30914b3a2e : SUCCESS in 3m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e79be7ac45514fbcad117f9082c40456 : SUCCESS in 1h 41m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0478b89b67344bd296dc6c0ff776ebc9 : SUCCESS in 3m 33s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5770d3a657cd428091906462ecc4f258 : SUCCESS in 1h 05m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c4aff9aceb4242a38f0938abb30340a4 : SUCCESS in 1h 09m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8022ac2069e5471d9797da92c22a496e : SUCCESS in 1h 10m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/87b625e2603641c18389867f03cfdfed : SUCCESS in 1h 01m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f5577e929c2b4793acd9bac7de534203 : SUCCESS in 27m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4692a21419fb48dab46492f7532033f5 : SUCCESS in 44m 37s","accounts_in_message":[],"_revision_number":3},{"id":"bb414a4d2633d79dc51d280726ff9c77d6e2e761","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-03 09:25:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"65b245aebcef3fecf0bc3e0b07d37b974401a8ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 09:26:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6892c0d779b1efd17b82bdedb4fc8ce3ccf7e930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 11:52:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab58a2846e884c0fb07de7b530fe37e7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7e6e286757c445919c8b8fd415d5c70c : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e401826550c42efb58a95843c4c1547 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c06a066c233d4545b02287643844ffa6 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27e5d5a716c24a46b944be9751b1aa15 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efb6a3afcff044b3a6d75806ad02dc0e : SUCCESS in 3m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/acc055d0e71444b18eab136de6db0d51 : SUCCESS in 2h 24m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c6ebee8b0e04059b6fd5c6525dd009a : SUCCESS in 3m 13s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f8bbb2485725496293a1cae69e87fdc4 : SUCCESS in 1h 06m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d5ffdb7ddeec42ffa3750a1559f75f70 : SUCCESS in 1h 03m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/850ba46a1a224d80a54abfbd3cca2fc9 : SUCCESS in 1h 12m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/52ffa2476b1f4cd1bace48328ed81e54 : SUCCESS in 1h 07m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a3f54d4739ce4ae6bd00bf45d3c816d5 : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":3},{"id":"05e89f754d4b231fc18531653a168c8c5aa933e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 11:52:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2ced2460ab672032b2011fa16144ccd63e5aa532","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 11:53:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2643cf21ed1446b496a3fbbca4c19dba\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/98bd87831f1a499eb6a162e31d80a494 : SUCCESS in 48s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dce69c41dbee454e82cf6fe0b6eea747 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b17bb4f65b16406bb12dd1a1553cdc18 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3169bf9f334be7d57722430c3f15b16771ee86b4","revisions":{"45653ff36480201ba482d9d47cfbd306ffbac2c2":{"kind":"REWORK","_number":1,"created":"2023-01-13 11:52:35.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/80/870080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/80/870080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/870080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/870080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/870080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/80/870080/1"}}},"commit":{"parents":[{"commit":"f2bb7c72cd1fa8bfd25431834dde6dbfcf4d0b72","subject":"\"convert_to_sanitized_mac_address\" accepts netaddr.EUI type values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f2bb7c72cd1fa8bfd25431834dde6dbfcf4d0b72"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2022-12-28 14:57:58.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-01-09 16:01:37.000000000","tz":60},"subject":"WIP port-hints: api definition","message":"WIP port-hints: api definition\n\nTODO api-ref\nTODO unit tests\nTODO relnote\n\nChange-Id: I5cb21b25fcda62eae3743b2f96177dcc1fda4afb\nPartially-Implements: #1990842\nRelated-Change (spec): https://review.opendev.org/c/openstack/neutron-specs/+/862133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/45653ff36480201ba482d9d47cfbd306ffbac2c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/45653ff36480201ba482d9d47cfbd306ffbac2c2"}]},"branch":"refs/heads/master"},"4d0fbdeb2956da40cdd55fbe51b0f7e7402ef24b":{"kind":"REWORK","_number":2,"created":"2023-01-18 13:40:48.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/80/870080/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/80/870080/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/870080/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/870080/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/870080/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/80/870080/2"}}},"commit":{"parents":[{"commit":"2ce390093be7bb722feddc81a770d878ede15dfd","subject":"Merge \"Update envlist for tox4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2ce390093be7bb722feddc81a770d878ede15dfd"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2022-12-28 14:57:58.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-01-18 12:58:41.000000000","tz":60},"subject":"port-hints: api definition","message":"port-hints: api definition\n\nThe port hints extension (``port-hints``) introduces the ``hints``\nport attribute. Hints are backend specific pieces of information,\nmainly to allow backend specific performance tuning. In itself this\nextension defines no particular hint, and therefore no valid values of\nthe ``hints`` attribute. It just serves as the base for other extensions\nintroducing concrete hints and signals the presence of the ``hints``\nport attribute to the API user. By default policy, use of the ``hints``\nattribute is restricted to admininstrative users.\n\nChange-Id: I5cb21b25fcda62eae3743b2f96177dcc1fda4afb\nPartially-Implements: #1990842\nRelated-Change (spec): https://review.opendev.org/c/openstack/neutron-specs/+/862133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4d0fbdeb2956da40cdd55fbe51b0f7e7402ef24b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4d0fbdeb2956da40cdd55fbe51b0f7e7402ef24b"}]},"branch":"refs/heads/master"},"3169bf9f334be7d57722430c3f15b16771ee86b4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-02 10:24:29.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/80/870080/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/80/870080/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/870080/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/870080/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/870080/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/80/870080/3"}}},"commit":{"parents":[{"commit":"b1a4f783daa1c48057cb450e58198f1bf93931c6","subject":"Fix pep8 errors with pytlint v2.16.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b1a4f783daa1c48057cb450e58198f1bf93931c6"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2022-12-28 14:57:58.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-02-02 10:23:38.000000000","tz":60},"subject":"port-hints: api definition","message":"port-hints: api definition\n\nThe port hints extension (``port-hints``) introduces the ``hints``\nport attribute. Hints are backend specific pieces of information,\nmainly to allow backend specific performance tuning. In itself this\nextension defines no particular hint, and therefore no valid values of\nthe ``hints`` attribute. It just serves as the base for other extensions\nintroducing concrete hints and signals the presence of the ``hints``\nport attribute to the API user. By default policy, use of the ``hints``\nattribute is restricted to admininstrative users.\n\nChange-Id: I5cb21b25fcda62eae3743b2f96177dcc1fda4afb\nPartially-Implements: #1990842\nRelated-Change (spec): https://review.opendev.org/c/openstack/neutron-specs/+/862133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3169bf9f334be7d57722430c3f15b16771ee86b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3169bf9f334be7d57722430c3f15b16771ee86b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
